)]}'
{"id":"openstack%2Fpuppet-ceph~363164","triplet_id":"openstack%2Fpuppet-ceph~master~Idea369a3f08be2d6c25913152b63cab5eadb8641","project":"openstack/puppet-ceph","branch":"master","topic":"selinux","hashtags":[],"change_id":"Idea369a3f08be2d6c25913152b63cab5eadb8641","subject":"Make RGW deployment idempotent","status":"MERGED","created":"2016-08-30 19:41:25.000000000","updated":"2016-09-01 17:31:34.000000000","submitted":"2016-09-01 16:49:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"096c57da4ce76e1aa3de85c09cdc375b73b51cf6","_number":363164,"virtual_id_number":363164,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2016-09-01 16:49:23.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2016-09-01 17:31:34.000000000","post_submit":true,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"all":[{"value":2,"date":"2016-09-01 16:12:57.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-08-31 18:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2016-08-31 11:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"all":[{"value":1,"date":"2016-09-01 16:12:57.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-08-30 19:53:08.000000000","updated_by":{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"},"reviewer":{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"},"state":"REVIEWER"},{"updated":"2016-08-31 11:37:13.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2016-08-31 18:39:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-09-01 16:12:57.000000000","updated_by":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"reviewer":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"state":"REVIEWER"},{"updated":"2016-09-01 16:49:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-09-01 17:31:34.000000000","updated_by":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c77afb5d55f4e3de6ab1e14c791fa3bdd66ae4c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-08-30 19:41:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91ce32ae2daeb03c9f5b1efa3730a5a4fb4e4637","author":{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"},"date":"2016-08-30 19:53:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8dee32c596d763849f5865164601f87f3883454","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 21:21:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-lint/0248ffa/ : SUCCESS in 2m 43s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-syntax-3-centos-7/d4e766c/ : FAILURE in 13m 02s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-syntax-4-centos-7/03bd1b1/ : SUCCESS in 3m 50s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/4678aa5/ : SUCCESS in 5m 38s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/2f0dffb/ : SUCCESS in 5m 43s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/5f3974a/ : FAILURE in 1h 05m 11s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-latest-centos-7/fd9f446/ : SUCCESS in 6m 19s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/eb300ee/ : SUCCESS in 5m 47s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/54e836b/ : SUCCESS in 4m 48s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/d40d4dd/ : SUCCESS in 25m 00s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/eafa3ce/ : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":2},{"id":"0f9129767a48c0b50951e2199e6efd856e329f1f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-08-30 21:28:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9cfa8b4caa7c6cfc255484115a97cc2dfe8eb140","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-30 21:36:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2338/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4491/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4428/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e762a9c3d2c5ef22dd2c1469382e1c8f75bb6af7","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-30 21:52:04.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2339/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4492/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4429/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"acd160b2c13df791bcc30b75f443afc260a6c9e0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 22:01:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-lint/59e4b6a/ : SUCCESS in 3m 28s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-syntax-3-centos-7/355e574/ : SUCCESS in 2m 55s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-syntax-4-centos-7/95252b7/ : SUCCESS in 3m 01s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/4a035d5/ : SUCCESS in 5m 03s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/8abfc27/ : FAILURE in 4m 41s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/e4eca46/ : SUCCESS in 9m 56s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-latest-centos-7/b80c505/ : SUCCESS in 6m 33s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/91a4a7d/ : SUCCESS in 5m 37s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/99e3d8c/ : SUCCESS in 6m 02s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/980d967/ : SUCCESS in 32m 39s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/ff41347/ : SUCCESS in 28m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5a27f1d976ec26978f3cd1074f8ad2f4f122874e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-08-30 23:14:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eee468a407372243efbf4ecc78e2d4619c753312","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-30 23:16:45.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2341/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4494/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4431/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"88e750430f48d0870ceed8d4542f1ba8e900d8a9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 23:48:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-lint/8fc440b/ : SUCCESS in 2m 43s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-syntax-3-centos-7/5fe5736/ : SUCCESS in 6m 55s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-syntax-4-centos-7/fa91f77/ : FAILURE in 12m 07s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/acde9c0/ : SUCCESS in 6m 36s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/b78d57a/ : SUCCESS in 6m 36s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/1901cd7/ : SUCCESS in 5m 26s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-unit-latest-centos-7/a5ab19b/ : SUCCESS in 5m 23s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/e6de29a/ : SUCCESS in 5m 35s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/2/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/7968fcb/ : SUCCESS in 5m 08s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/bd5e2e3/ : SUCCESS in 27m 13s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/2/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/6a600a0/ : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":2},{"id":"14952116f5d915e977c1398224739ee79a5bcd40","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 01:22:41.000000000","message":"Patch Set 2:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/64/363164/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/a3c763c/ : POST_FAILURE in 2h 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4381972a111cbe7512545236c2fb350eeec7f773","author":{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"},"date":"2016-08-31 03:59:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a98e1c16414f434d8ef9d458b04850b1abe32241","author":{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"},"date":"2016-08-31 04:07:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"25bfa33326e9173888bccd90826cf2affc71ffbf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 04:38:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-lint/70288db/ : SUCCESS in 2m 33s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-3-centos-7/db8f131/ : SUCCESS in 2m 53s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-4-centos-7/a953d9e/ : SUCCESS in 3m 18s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/ea26306/ : SUCCESS in 7m 05s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/9d8fbc5/ : FAILURE in 5m 07s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/007855f/ : SUCCESS in 6m 10s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-latest-centos-7/1415332/ : SUCCESS in 8m 56s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/1809f67/ : SUCCESS in 6m 45s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/40fc121/ : SUCCESS in 5m 57s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/15e1ab3/ : SUCCESS in 32m 18s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/98dec63/ : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":3},{"id":"cc1f4e3a7e208c8a5436f9cc5624aaa69a1bf236","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-31 05:51:47.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2347/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4500/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4437/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"d8771c5ecb71cf6f08c2bc007afc1774eb379822","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-31 06:08:57.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2348/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4501/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4438/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"dddcbf8eb7945cb6ba304eef03b76b824551d0b8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 06:56:41.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/64/363164/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/1ffc895/ : POST_FAILURE in 2h 49m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7536155ee020ee0ee528a07227071060b379e13a","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2016-08-31 10:56:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"61810fc4928d2472fdb6d57a7b1ab818ad56063e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 11:31:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-lint/fe6735c/ : SUCCESS in 2m 35s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-3-centos-7/7579dbe/ : SUCCESS in 3m 14s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-4-centos-7/f206b9e/ : SUCCESS in 5m 13s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/f5bf36c/ : SUCCESS in 5m 16s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/1b81eaf/ : SUCCESS in 7m 28s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/c7d3603/ : SUCCESS in 6m 44s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-latest-centos-7/ff26778/ : SUCCESS in 5m 06s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/f8c4273/ : SUCCESS in 6m 07s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/26c78c1/ : SUCCESS in 5m 10s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/68e94f7/ : SUCCESS in 30m 00s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/3e727ba/ : SUCCESS in 27m 07s","accounts_in_message":[],"_revision_number":3},{"id":"1a1447d06c000abbdfbfe9fcb81669462548931c","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2016-08-31 11:37:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9c327876134141a193180e5e9c2f289118973ff8","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-31 12:31:14.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2356/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4509/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4446/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"dfb4bfa3778b645ef4efd434bf63d6ecb6b1585e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-08-31 12:46:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"abc6933c50d0f7c48d3283cb5be3aa5bf818bfe5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 13:20:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-lint/d7858e1/ : SUCCESS in 3m 58s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-3-centos-7/fa8ac86/ : SUCCESS in 3m 24s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-4-centos-7/bc89f01/ : SUCCESS in 5m 11s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/5bc7f61/ : SUCCESS in 5m 16s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/4f4f9ef/ : SUCCESS in 5m 35s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/fc7ef87/ : SUCCESS in 10m 24s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-latest-centos-7/03d3a9f/ : SUCCESS in 6m 53s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/40e697e/ : SUCCESS in 11m 35s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/eaa5402/ : SUCCESS in 10m 24s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/bfbf02e/ : SUCCESS in 23m 15s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/29232f5/ : SUCCESS in 28m 37s","accounts_in_message":[],"_revision_number":3},{"id":"1107c9978725e4a372a8237baec2d4684420da95","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-08-31 13:47:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"51133b9aad78f0d3706ddd413d5700382d85849a","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-08-31 13:49:34.000000000","message":"Patch Set 3:\n\nfuel: recheck","accounts_in_message":[],"_revision_number":3},{"id":"90ff420ac7cdc3174725558b13876fa697476762","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 14:24:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-lint/e9d782f/ : SUCCESS in 2m 59s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-3-centos-7/04781d6/ : SUCCESS in 3m 08s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-4-centos-7/f56a971/ : SUCCESS in 4m 20s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/4349665/ : SUCCESS in 7m 45s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/a03f11d/ : SUCCESS in 7m 49s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/ce17c52/ : SUCCESS in 10m 18s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-latest-centos-7/534b624/ : SUCCESS in 7m 26s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/9a6ba2e/ : SUCCESS in 9m 55s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/e5def3e/ : SUCCESS in 5m 28s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/08ade89/ : SUCCESS in 23m 28s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/43107db/ : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e3d38ad8c08abe0976e9c37cf334aef34d65420a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 14:25:28.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/64/363164/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/2174190/ : FAILURE in 2h 03m 32s","accounts_in_message":[],"_revision_number":3},{"id":"9465bb940a9ab5928e0d9ab4edbcb86ccec03435","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-31 14:33:55.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2362/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4515/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4452/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"62639868e8619cba3d28f4286cde61263bfe73ba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-08-31 18:39:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0bf7062ff263404fddaf3ae6b94985af425cd6a5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-08-31 18:46:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9f2d07afc47d4f9ec6f8f7ed2bec53f9a9cca50f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 19:44:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-lint/cad0cfe/ : SUCCESS in 3m 05s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-3-centos-7/debba2e/ : SUCCESS in 3m 25s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-syntax-4-centos-7/b7ee653/ : SUCCESS in 3m 16s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.6-centos-7/8bc999c/ : SUCCESS in 7m 48s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-3.8-centos-7/21eb81f/ : SUCCESS in 8m 04s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-4.5-centos-7/0c87299/ : SUCCESS in 5m 40s\n- gate-puppet-ceph-puppet-unit-latest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-unit-latest-centos-7/21a4fc6/ : SUCCESS in 5m 01s (non-voting)\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-centos-7/2733a96/ : SUCCESS in 6m 04s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/f216df4/ : SUCCESS in 5m 43s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/db82bee/ : SUCCESS in 23m 18s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/3/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/fcf6f24/ : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":3},{"id":"0cb7d379b5c986930cac6d647c59dbb9618bb218","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-31 20:36:04.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2369/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4522/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4459/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c5c36df4b8b4d20052997376431a17f9af79b607","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-09-01 11:16:50.000000000","message":"Patch Set 3:\n\nfuel: recheck","accounts_in_message":[],"_revision_number":3},{"id":"dc923487ad7d6ed10793316e4bb87f7628a7c62c","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-09-01 13:25:16.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2393/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4546/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4483/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2f00794ba79cf21ea3b93be348a84ff14399f30d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-01 15:29:45.000000000","message":"Patch Set 3:\n\nfuel: recheck","accounts_in_message":[],"_revision_number":3},{"id":"17cfad492794304244b990e0273934530f44e479","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-09-01 16:12:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ed3d6959f7350a2ba408687556d98b0f3e85390b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-01 16:13:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"947a16e71a11480d6ab27231446113f043b060c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-01 16:49:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-ceph-puppet-lint http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-lint/df24fc8/ : SUCCESS in 3m 37s\n- gate-puppet-ceph-puppet-syntax-3-centos-7 http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-syntax-3-centos-7/f440faa/ : SUCCESS in 5m 09s\n- gate-puppet-ceph-puppet-syntax-4-centos-7 http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-syntax-4-centos-7/3edcefa/ : SUCCESS in 3m 08s\n- gate-puppet-ceph-puppet-unit-3.6-centos-7 http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-unit-3.6-centos-7/97fe11d/ : SUCCESS in 5m 34s\n- gate-puppet-ceph-puppet-unit-3.8-centos-7 http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-unit-3.8-centos-7/7738c0c/ : SUCCESS in 6m 30s\n- gate-puppet-ceph-puppet-unit-4.5-centos-7 http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-unit-4.5-centos-7/d022932/ : SUCCESS in 5m 55s\n- gate-puppet-ceph-puppet-beaker-rspec-centos-7 http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-beaker-rspec-centos-7/1c6b8ab/ : SUCCESS in 6m 35s\n- gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/64/363164/3/gate/gate-puppet-ceph-puppet-beaker-rspec-ubuntu-xenial/2664a4f/ : SUCCESS in 6m 55s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/64/363164/3/gate/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/b2d1484/ : SUCCESS in 29m 31s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/64/363164/3/gate/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/10b2cb3/ : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9c72ebf6b8ee11cb43356620796968724e4b3d78","date":"2016-09-01 16:49:24.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"81760227a6128fab3a0db90af21b50db4813708e","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-09-01 17:31:34.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2404/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4557/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4494/ : SUCCESS","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b72bf9d6b8fde1feba4c4e7ef5c38249d58c3413","revisions":{"cf43e9b30c203d92f41ad519e007e4d978506641":{"kind":"REWORK","_number":1,"created":"2016-08-30 19:41:25.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/64/363164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ceph","ref":"refs/changes/64/363164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/1"}}},"commit":{"parents":[{"commit":"309ef4819b4c1ddbcc83e43fc931a2135882fa68","subject":"Fix the logic to skip SELinux labeling of block devices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/309ef4819b4c1ddbcc83e43fc931a2135882fa68"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-08-30 19:38:58.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-08-30 19:41:24.000000000","tz":-240},"subject":"Make RGW deployment idempotent","message":"Make RGW deployment idempotent\n\n- Set selinux_ignore_defaults to True for RGW directories otherwise Puppet\n  will try to modify SElinux context, different from packaging.\n- Run systemctl-reload-from-rgw only once, when package is installed.\n\nChange-Id: Idea369a3f08be2d6c25913152b63cab5eadb8641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/cf43e9b30c203d92f41ad519e007e4d978506641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/cf43e9b30c203d92f41ad519e007e4d978506641"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"309ef4819b4c1ddbcc83e43fc931a2135882fa68","is_merged_in_target_branch":true,"change_id":"I4ab32e5f796609cf2f985e989c6dc04d0eb37190","change_number":361159,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a37c8dbd6c060884737388590482fd339df3a855":{"kind":"REWORK","_number":2,"created":"2016-08-30 19:53:08.000000000","uploader":{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"},"ref":"refs/changes/64/363164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ceph","ref":"refs/changes/64/363164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/2"}}},"commit":{"parents":[{"commit":"309ef4819b4c1ddbcc83e43fc931a2135882fa68","subject":"Fix the logic to skip SELinux labeling of block devices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/309ef4819b4c1ddbcc83e43fc931a2135882fa68"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-08-30 19:38:58.000000000","tz":-240},"committer":{"name":"Keith Schincke","email":"keith.schincke@gmail.com","date":"2016-08-30 19:52:34.000000000","tz":-240},"subject":"Make RGW deployment idempotent","message":"Make RGW deployment idempotent\n\n- Set selinux_ignore_defaults to True for RGW directories otherwise Puppet\n  will try to modify SElinux context, different from packaging.\n\nChange-Id: Idea369a3f08be2d6c25913152b63cab5eadb8641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/a37c8dbd6c060884737388590482fd339df3a855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/a37c8dbd6c060884737388590482fd339df3a855"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"309ef4819b4c1ddbcc83e43fc931a2135882fa68","is_merged_in_target_branch":true,"change_id":"I4ab32e5f796609cf2f985e989c6dc04d0eb37190","change_number":361159,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b72bf9d6b8fde1feba4c4e7ef5c38249d58c3413":{"kind":"REWORK","_number":3,"created":"2016-08-31 03:59:11.000000000","uploader":{"_account_id":19564,"name":"Keith Schincke","email":"keith.schincke@gmail.com","username":"colonwq"},"ref":"refs/changes/64/363164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ceph","ref":"refs/changes/64/363164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ceph refs/changes/64/363164/3"}}},"commit":{"parents":[{"commit":"309ef4819b4c1ddbcc83e43fc931a2135882fa68","subject":"Fix the logic to skip SELinux labeling of block devices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/309ef4819b4c1ddbcc83e43fc931a2135882fa68"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-08-30 19:38:58.000000000","tz":-240},"committer":{"name":"Keith Schincke","email":"keith.schincke@gmail.com","date":"2016-08-31 03:58:41.000000000","tz":-240},"subject":"Make RGW deployment idempotent","message":"Make RGW deployment idempotent\n\n- Set selinux_ignore_defaults to True for RGW directories otherwise Puppet\n  will try to modify SElinux context, different from packaging.\n- Update rgw service for systemd\n\nChange-Id: Idea369a3f08be2d6c25913152b63cab5eadb8641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/b72bf9d6b8fde1feba4c4e7ef5c38249d58c3413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/b72bf9d6b8fde1feba4c4e7ef5c38249d58c3413"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"309ef4819b4c1ddbcc83e43fc931a2135882fa68","is_merged_in_target_branch":true,"change_id":"I4ab32e5f796609cf2f985e989c6dc04d0eb37190","change_number":361159,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
