)]}'
{"id":"openstack%2Fpuppet-ceph~543319","triplet_id":"openstack%2Fpuppet-ceph~master~Id8b37831c20123dcd69ce672dadc48c00a2e0524","project":"openstack/puppet-ceph","branch":"master","topic":"gemfile-fix","hashtags":[],"change_id":"Id8b37831c20123dcd69ce672dadc48c00a2e0524","subject":"Load puppet-openstack_spec_helper locally during tests","status":"MERGED","created":"2018-02-12 02:55:45.000000000","updated":"2018-02-13 00:12:22.000000000","submitted":"2018-02-13 00:12:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"543319-1518480742140-0cfd7981","meta_rev_id":"04b507716e2fe510bb609c50f83f63332df8bcfb","_number":543319,"virtual_id_number":543319,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2018-02-12 04:53:46.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2018-02-13 00:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2018-02-12 14:54:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2018-02-12 15:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2018-02-12 15:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-12 14:54:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-02-12 15:15:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2018-02-13 00:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"542a563753000e9b826098fdd475f8472a5d627c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-02-12 02:55:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ad881931eb32dc82b22d903941dc7abf652c285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-12 04:00:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/19/543319/1/check/puppet-openstack-lint/7a73ec3/ : SUCCESS in 3m 11s\n- puppet-openstack-syntax-4 http://logs.openstack.org/19/543319/1/check/puppet-openstack-syntax-4/7348b5b/ : SUCCESS in 4m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-unit-4.8-centos-7/a28d9f0/ : SUCCESS in 10m 43s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-unit-latest-centos-7/8ed4e96/ : FAILURE in 7m 08s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/224fca6/ : FAILURE in 11m 46s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/d4d1c8e/ : FAILURE in 57m 53s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/1778f6e/ : FAILURE in 28m 16s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/bbbefa5/ : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":1},{"id":"91c9fd22d19dde1eccf0201b5eb0bdc55c0286db","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-02-12 04:53:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e29d4f48a326bbf01de991ce83eaa96951c4de3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-12 06:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/19/543319/1/check/puppet-openstack-lint/1b920ee/ : SUCCESS in 3m 14s\n- puppet-openstack-syntax-4 http://logs.openstack.org/19/543319/1/check/puppet-openstack-syntax-4/0b47da7/ : SUCCESS in 3m 41s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-unit-4.8-centos-7/5f60d0f/ : SUCCESS in 11m 33s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-unit-latest-centos-7/6c5db7c/ : FAILURE in 7m 46s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/784b989/ : FAILURE in 12m 27s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/0a8d9d4/ : SUCCESS in 1h 28m 16s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/87e3832/ : SUCCESS in 32m 51s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/85b949a/ : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9bf026c07ac4dd89f53da647d6e79ad851a2f79b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-02-12 14:54:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"342fe24b3cd52364392e59f5a2d44debf5364fdd","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-02-12 15:15:55.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d3ec494f81eed98c68a1c07c43ce8a3b287f64a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-12 15:16:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fd25a8821504c21a4a1cfd96b601fe212f9abdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-12 16:37:53.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"229319c57cff3916c5696d8b2d0aaf9ff6c70a39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-12 17:46:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/19/543319/1/gate/puppet-openstack-lint/79e37ca/ : SUCCESS in 4m 24s\n- puppet-openstack-syntax-4 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-syntax-4/eb5358e/ : SUCCESS in 4m 40s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-unit-4.8-centos-7/cd27c0a/ : SUCCESS in 10m 10s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/8decc5a/ : FAILURE in 58m 14s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/08c55c6/ : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f4c91ae96bbbe0db01f498c97c59fd9bc7ae6519","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-02-12 18:50:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"233bf8a5a5202502f938e6663bf9479d42be0971","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-12 22:45:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/19/543319/1/check/puppet-openstack-lint/849d95d/ : SUCCESS in 3m 02s\n- puppet-openstack-syntax-4 http://logs.openstack.org/19/543319/1/check/puppet-openstack-syntax-4/56ea91b/ : SUCCESS in 3m 00s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-unit-4.8-centos-7/d8f2cdc/ : SUCCESS in 10m 06s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-unit-latest-centos-7/bff5c16/ : FAILURE in 8m 58s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/7c52796/ : FAILURE in 13m 10s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/dbc6dca/ : SUCCESS in 1h 02m 08s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/6ffa099/ : SUCCESS in 29m 48s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/543319/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/af268ba/ : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":1},{"id":"06dfd6b4f406201edc2186b9322b3b5ff5d4b20c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-12 22:56:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"920d23b36c0eff091f563dbc9cd09d4e7483a3c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 00:12:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/19/543319/1/gate/puppet-openstack-lint/127ea1e/ : SUCCESS in 3m 13s\n- puppet-openstack-syntax-4 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-syntax-4/4401d92/ : SUCCESS in 3m 08s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-unit-4.8-centos-7/2f3b8a3/ : SUCCESS in 9m 36s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/91c7d92/ : SUCCESS in 1h 09m 30s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/543319/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/c0c5fee/ : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":1},{"id":"474e0235d09f196ba6c0c692f6c4658ab4031024","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 00:12:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"beb527799ab6ef205e8e22b91890f657a95f8965","revisions":{"beb527799ab6ef205e8e22b91890f657a95f8965":{"kind":"REWORK","_number":1,"created":"2018-02-12 02:55:45.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/19/543319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ceph","ref":"refs/changes/19/543319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/19/543319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/19/543319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ceph refs/changes/19/543319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ceph refs/changes/19/543319/1"}}},"commit":{"parents":[{"commit":"2b8bbde9df08509572bd6400ecb374e091acf118","subject":"Merge \"Zuul: Remove project name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/2b8bbde9df08509572bd6400ecb374e091acf118"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2018-02-12 02:53:34.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2018-02-12 02:53:34.000000000","tz":480},"subject":"Load puppet-openstack_spec_helper locally during tests","message":"Load puppet-openstack_spec_helper locally during tests\n\nChange-Id: Id8b37831c20123dcd69ce672dadc48c00a2e0524\nCo-Authored-By: Mohammed Naser \u003cmnaser@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/beb527799ab6ef205e8e22b91890f657a95f8965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceph/commit/beb527799ab6ef205e8e22b91890f657a95f8965"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b8bbde9df08509572bd6400ecb374e091acf118","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
