)]}'
{"id":"openstack%2Fpuppet-cinder~1002587","triplet_id":"openstack%2Fpuppet-cinder~master~If53fca5d20912ff164fb6f74f478b24b769738e9","project":"openstack/puppet-cinder","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-27 08:19:41.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If53fca5d20912ff164fb6f74f478b24b769738e9","subject":"Install optional dependencies for backup driver","status":"NEW","created":"2026-08-27 06:46:06.000000000","updated":"2026-08-31 07:33:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"81f5e52caf618fdba836c419b419f085a1263f22","_number":1002587,"virtual_id_number":1002587,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-31 07:33:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-27 08:19:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2aaf6b20e82abda6118e8d9697a2680ffb1a479e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-27 06:46:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7f70555b388a1c35f2c0ddb239878413f7cc6ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 08:19:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57a6d7e150984012ba96aab5822f9a6a\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/0f9971e37af04075a40556cad5884693 : SUCCESS in 2m 23s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/898a97deee454d91b271041fc05e403f : FAILURE in 2m 22s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28af0ae0deb04dab9498c6a15c7b6a4d : SUCCESS in 3m 09s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/ac92ca9c439b4841bf16fda08851062e : FAILURE in 4m 59s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ab529e1fdc9412e88efed836c5a7759 : FAILURE in 5m 09s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74e7da5f36934ce1927a1e89f685f009 : FAILURE in 5m 23s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d91ecdf9da4237a832199fb16fa852 : SUCCESS in 1h 04m 35s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/195b62562a5f4efcaf881bf588fa5935 : FAILURE in 43m 35s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59ea2a368fd94a709edde8209569e0bc : FAILURE in 1h 20m 07s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09b657f96bff4bdc89d993a8345899ab : SUCCESS in 50m 08s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44e44fd1686f44aa8bec1dd5d952c530 : FAILURE in 48m 15s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb2aaa464760472c9aa3b6bb5f7d6a94 : SUCCESS in 1h 02m 13s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/546fb161b130495a8a3109d220747fea : SUCCESS in 1h 10m 59s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80673f547a0849fbbae5ebd1f789f32f : SUCCESS in 59m 53s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7655dfe14f1f41cfb88247474353ada5 : SUCCESS in 1h 01m 20s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c12a69b3c4514d30b154e74bbadca7e4 : SUCCESS in 54m 59s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d85b01e314da427a8be5a7a4474bd80e : SUCCESS in 1h 05m 42s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/65292142a14c42f2a6bcdc574d27a1c9 : SUCCESS in 1h 26m 37s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70d00298de474dce8b9a61bd1629b951 : SUCCESS in 1h 00m 22s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac37081a036a4681b5f56f544d32bb95 : FAILURE in 8m 24s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/55303dcfb39740e39a54a1219ca976b1 : SUCCESS in 59m 00s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c66ebdab55a1477b9ef3cda9a45fca2b : SUCCESS in 22m 52s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb3ae149287f40719848448e6168ac07 : SUCCESS in 8m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a090913d9fbc4cba86a6c582bd6fb938 : SUCCESS in 8m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d66b9db2d79541caa3b1f64aa7d362d9 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"243db706b9f8c6a5c2421169bf8db4a6c2a9041f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 06:04:08.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"dfd6ce2f907c147429fe54793f5c496f073aa467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 06:09:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"81f5e52caf618fdba836c419b419f085a1263f22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 07:33:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1033da1e9b04acfa6fbba7863811930\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2c7e11d46bae429ca08ac03c3ea567e8 : SUCCESS in 1m 29s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d31889d0383a4a408fc00722ccecd292 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5d2876ff4d44639afeef9add9755d1f : SUCCESS in 2m 47s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/cda2a626a3ff41589769260624e871c7 : FAILURE in 4m 57s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b507e676e6d8461182cda3700ab9bbfb : FAILURE in 8m 12s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72fac56756ed4080b957db4cd5868d4a : FAILURE in 7m 37s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a40c6e08c3b540a3b35803e5e4757d6c : SUCCESS in 54m 22s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7cb1f072853461e95c9f161e87982a9 : FAILURE in 1h 12m 23s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/154b064e82334e0b83943723bfeebfd1 : FAILURE in 1h 18m 24s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e20cb957d27540e594278af3788a69f1 : SUCCESS in 35m 02s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc2fc8ed6a4a4a52a2672c7d9a7321b1 : FAILURE in 56m 20s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e4d6222909540408db01ce15482ac79 : SUCCESS in 1h 03m 12s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e869ebb232e0464b9779127ff47c13d2 : SUCCESS in 1h 22m 22s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a805c761c9be42c6b3a008e68c1484ae : SUCCESS in 57m 46s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9c9b9f4d6fe41f9a4f31a2880d2dc44 : SUCCESS in 56m 59s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/636c71ffacfb4c1981368dce6bede505 : SUCCESS in 59m 27s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cf8d176a617415593c1e3cd8211a7e1 : SUCCESS in 1h 04m 21s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63f53040b5bc4b5282279604ae067f84 : SUCCESS in 1h 20m 27s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc82c564c44847d79f1d428006fd1736 : SUCCESS in 1h 02m 56s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/582f122fe0c644999954baef6d5c763f : SUCCESS in 55m 18s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abed4c6ff2d54d4c9832a7bb69e6eba9 : SUCCESS in 58m 40s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/facac8c04aad4fffb6ed0a247e56ca79 : SUCCESS in 19m 24s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67f4c3914cb64e1db606f75f320010aa : SUCCESS in 13m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a55f867501240af9d7a81853f4eedf7 : SUCCESS in 10m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82fe9246352a4196ae53df3c9e9635e8 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"39167aff1988d1c2bbb9e121929381482910b1c3","revisions":{"17aacd99a24e7840b7fcabe359d9add5890c137c":{"kind":"REWORK","_number":1,"created":"2026-08-27 06:46:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/1002587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/87/1002587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/1"}}},"commit":{"parents":[{"commit":"f623d03898159fbf8228388a7c489075f638f5d2","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/f623d03898159fbf8228388a7c489075f638f5d2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 06:44:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 06:45:52.000000000","tz":540},"subject":"Install optional dependencies for backup driver","message":"Install optional dependencies for backup driver\n\nLatest cinder no longer define these as the mandatory dependencies and\ndeployment tools should install these explicitly when needed.\n\nChange-Id: If53fca5d20912ff164fb6f74f478b24b769738e9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/17aacd99a24e7840b7fcabe359d9add5890c137c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/17aacd99a24e7840b7fcabe359d9add5890c137c"}]},"branch":"refs/heads/master"},"60418dbc32ba3884580b62996376cbcc3e4630b2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-31 06:04:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/1002587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/87/1002587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/2"}}},"commit":{"parents":[{"commit":"f623d03898159fbf8228388a7c489075f638f5d2","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/f623d03898159fbf8228388a7c489075f638f5d2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 06:44:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-31 06:04:08.000000000","tz":0},"subject":"Install optional dependencies for backup driver","message":"Install optional dependencies for backup driver\n\nLatest cinder no longer define these as the mandatory dependencies and\ndeployment tools should install these explicitly when needed.\n\nDepends-on: https://review.opendev.org/926868\nChange-Id: If53fca5d20912ff164fb6f74f478b24b769738e9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/60418dbc32ba3884580b62996376cbcc3e4630b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/60418dbc32ba3884580b62996376cbcc3e4630b2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"39167aff1988d1c2bbb9e121929381482910b1c3":{"kind":"REWORK","_number":3,"created":"2026-08-31 06:09:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/1002587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/87/1002587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/87/1002587/3"}}},"commit":{"parents":[{"commit":"f623d03898159fbf8228388a7c489075f638f5d2","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/f623d03898159fbf8228388a7c489075f638f5d2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 06:44:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-31 06:08:53.000000000","tz":540},"subject":"Install optional dependencies for backup driver","message":"Install optional dependencies for backup driver\n\nLatest cinder no longer define these as the mandatory dependencies and\ndeployment tools should install these explicitly when needed.\n\nDepends-on: https://review.opendev.org/926868\nChange-Id: If53fca5d20912ff164fb6f74f478b24b769738e9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/39167aff1988d1c2bbb9e121929381482910b1c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/39167aff1988d1c2bbb9e121929381482910b1c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
