)]}'
{"id":"openstack%2Fpuppet-cinder~223977","triplet_id":"openstack%2Fpuppet-cinder~stable%2Ficehouse~I964b86798b4462e7f2ed8729d6938eff162f1694","project":"openstack/puppet-cinder","branch":"stable/icehouse","topic":"bug/1482564","hashtags":[],"change_id":"I964b86798b4462e7f2ed8729d6938eff162f1694","subject":"Add an ability to manage use_stderr parameter","status":"MERGED","created":"2015-09-16 09:14:51.000000000","updated":"2015-09-21 10:30:37.000000000","submitted":"2015-09-21 10:30:36.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01515334b4a7f1694d8efd73f461a3cc33423298","_number":223977,"virtual_id_number":223977,"owner":{"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":0,"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":2,"date":"2015-09-21 10:30:36.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7604,"name":"Sergey Kolekonov","email":"skolekonov@mirantis.com","username":"skolekonov"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"all":[{"value":2,"date":"2015-09-21 10:25:41.000000000","_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":2,"date":"2015-09-21 09:32:51.000000000","_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7604,"name":"Sergey Kolekonov","email":"skolekonov@mirantis.com","username":"skolekonov"},{"value":1,"date":"2015-09-17 19:13:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"all":[{"value":1,"date":"2015-09-21 10:25:41.000000000","_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":0,"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-09-16 09:14:51.000000000","_account_id":7604,"name":"Sergey Kolekonov","email":"skolekonov@mirantis.com","username":"skolekonov"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2015-09-17 14:00:55.000000000","_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"_account_id":7604,"name":"Sergey Kolekonov","email":"skolekonov@mirantis.com","username":"skolekonov"},{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-09-16 09:14:51.000000000","updated_by":{"_account_id":7604,"name":"Sergey Kolekonov","email":"skolekonov@mirantis.com","username":"skolekonov"},"reviewer":{"_account_id":7604,"name":"Sergey Kolekonov","email":"skolekonov@mirantis.com","username":"skolekonov"},"state":"REVIEWER"},{"updated":"2015-09-17 19:13:41.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2015-09-21 09:32:51.000000000","updated_by":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"reviewer":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"state":"REVIEWER"},{"updated":"2015-09-21 10:25:41.000000000","updated_by":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"reviewer":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"state":"REVIEWER"},{"updated":"2015-09-21 10:30:36.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"3baf74ed57b357d19d7b40338c690df6d4ae41b6","author":{"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"},"date":"2015-09-16 09:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42b1d8ef7e73bb1d527894b6776761770c33c034","author":{"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"},"date":"2015-09-16 09:18:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"bf8f56e1085ffcb6962aa437e5bd95c75b29054c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-16 09:51:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-cinder-puppet-lint http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-lint/9896678/ : SUCCESS in 0s\n- gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7/efc975e/ : SUCCESS in 1m 07s\n- gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7/644b9d0/ : SUCCESS in 1m 06s\n- gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7/f3281e3/ : SUCCESS in 1m 11s\n- gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7/5a96e4b/ : SUCCESS in 1m 42s\n- gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7/ac5ea2d/ : SUCCESS in 2m 07s\n- gate-puppet-cinder-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-4.0-dsvm-centos7/cd53629/ : FAILURE in 1m 05s\n- gate-puppet-cinder-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-latest-dsvm-centos7/48d0f28/ : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d77963cc162f141726f063bcc3922e4a704483c7","author":{"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"},"date":"2015-09-16 14:51:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0509d0a3adfdd5506cf451687714a14d20539581","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-16 15:29:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-cinder-puppet-lint http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-lint/1ad8270/ : SUCCESS in 0s\n- gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7/636c562/ : SUCCESS in 42s\n- gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7/3749cf7/ : SUCCESS in 30s\n- gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7/46337d2/ : SUCCESS in 53s\n- gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7/36cd2d7/ : SUCCESS in 51s\n- gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7/e2e7b60/ : SUCCESS in 1m 03s\n- gate-puppet-cinder-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-4.0-dsvm-centos7/bb29ea1/ : FAILURE in 34s\n- gate-puppet-cinder-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-latest-dsvm-centos7/cf6d059/ : FAILURE in 53s","accounts_in_message":[],"_revision_number":1},{"id":"d02e3c35b2f41ad307a0b51171aac1ca2771c385","author":{"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"},"date":"2015-09-17 14:00:55.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bcd3fa1a9caa8a120d9593b69cc477d1a21893ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-17 14:43:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-cinder-puppet-lint http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-lint/adfc425/ : SUCCESS in 6s\n- gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7/55e54b8/ : SUCCESS in 1m 08s\n- gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7/429aa0d/ : SUCCESS in 1m 16s\n- gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7/d4ef5de/ : SUCCESS in 1m 16s\n- gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7/ae9d0c2/ : SUCCESS in 1m 13s\n- gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/77/223977/1/check/gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7/d4517df/ : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5673e51058e565560945dab8b098f19e5c719934","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2015-09-17 19:13:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"71835c3c26e8019b40ac2a536839d4fbb49b4499","author":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"date":"2015-09-21 09:32:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a01b0a95e28e2c41a788906d0edcc5f36e65bc5","author":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"date":"2015-09-21 10:25:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a31baee5113f812ce1d83a70b806ab8aaced35bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-21 10:25:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"d02165ae36dc979fc0d4f1ccd5ba0742321982d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-21 10:30:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-cinder-puppet-lint http://logs.openstack.org/77/223977/1/gate/gate-puppet-cinder-puppet-lint/8264c22/ : SUCCESS in 23s\n- gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/77/223977/1/gate/gate-puppet-cinder-puppet-unit-3.3-dsvm-centos7/5daa5c1/ : SUCCESS in 1m 08s\n- gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/77/223977/1/gate/gate-puppet-cinder-puppet-unit-3.4-dsvm-centos7/3324067/ : SUCCESS in 1m 02s\n- gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/77/223977/1/gate/gate-puppet-cinder-puppet-unit-3.6-dsvm-centos7/cc8df1c/ : SUCCESS in 1m 08s\n- gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/77/223977/1/gate/gate-puppet-cinder-puppet-unit-3.7-dsvm-centos7/7e419a0/ : SUCCESS in 1m 09s\n- gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/77/223977/1/gate/gate-puppet-cinder-puppet-unit-3.8-dsvm-centos7/17c086e/ : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d810f5b63175e753f64caaa5759d1cacca5931ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-21 10:30:37.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c351a699b69a4560253c0d934f15d76daaecd157","revisions":{"c351a699b69a4560253c0d934f15d76daaecd157":{"kind":"REWORK","_number":1,"created":"2015-09-16 09:14:51.000000000","uploader":{"_account_id":14027,"name":"Michal Rostecki","email":"mrostecki@suse.de","username":"mrostecki"},"ref":"refs/changes/77/223977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/77/223977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/77/223977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/77/223977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/77/223977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/77/223977/1"}}},"commit":{"parents":[{"commit":"a79ad481cf741e04caa44f230204cec2183ea6c3","subject":"Update .gitreview file for project rename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/a79ad481cf741e04caa44f230204cec2183ea6c3"}]}],"author":{"name":"Sergey Kolekonov","email":"skolekonov@mirantis.com","date":"2015-08-07 12:43:47.000000000","tz":180},"committer":{"name":"Michal Rostecki","email":"mrostecki@mirantis.com","date":"2015-09-16 09:14:12.000000000","tz":120},"subject":"Add an ability to manage use_stderr parameter","message":"Add an ability to manage use_stderr parameter\n\nWhen logging is already set up, it can be useful to disable this option\nto avoid logs duplication (collected by Upstart, for example)\n\nChange-Id: I964b86798b4462e7f2ed8729d6938eff162f1694\nCloses-bug: #1482564\n(cherry picked from commit c9de4988543b6f0bf55891a4f3f90fa804abe74b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/c351a699b69a4560253c0d934f15d76daaecd157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/c351a699b69a4560253c0d934f15d76daaecd157"}]},"parents_data":[{"branch_name":"refs/heads/stable/icehouse","commit_id":"a79ad481cf741e04caa44f230204cec2183ea6c3","is_merged_in_target_branch":true,"change_id":"I3b9c227101344175d5f27504049f9c7b1ff306d4","change_number":191427,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/icehouse"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
