)]}'
{"id":"openstack%2Fpuppet-cinder~376482","triplet_id":"openstack%2Fpuppet-cinder~master~I36db18a5a3bbcbfc6a80b9ec4dd1c772fd93cffa","project":"openstack/puppet-cinder","branch":"master","topic":"release/9.4.0","hashtags":[],"change_id":"I36db18a5a3bbcbfc6a80b9ec4dd1c772fd93cffa","subject":"Puppet OpenStack Newton RC2","status":"MERGED","created":"2016-09-26 13:32:09.000000000","updated":"2016-09-27 00:47:15.000000000","submitted":"2016-09-27 00:47:15.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0dbd502c2518cd3057088628a122183542936c3","_number":376482,"virtual_id_number":376482,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2016-09-27 00:47:15.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2016-09-26 13:37:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-09-26 14:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"value":1,"date":"2016-09-26 14:08:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2016-09-26 13:38:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":1,"date":"2016-09-26 16:40:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-09-26 13:37:08.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-09-26 14:07:02.000000000","updated_by":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"reviewer":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"state":"REVIEWER"},{"updated":"2016-09-26 14:08:05.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2016-09-26 16:40:53.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2016-09-27 00:47:15.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5d54610a5e177167ec2c1df8903b2d7da6109127","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-09-26 13:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690548bd0153f5fcd740002f3a2e237e185feb5d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-26 13:37:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6509aa0462266ffd63281a4b9d0210912697b1e","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2016-09-26 13:38:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a2320e52d6964b6245bcab52a0c27050c8a1f08","author":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"date":"2016-09-26 14:07:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"954d10f8dd54af8690395c351a3bafd4f0b0c7e5","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2016-09-26 14:08:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4b5642d6d7d864d2cf86c7e2988c509c8412c50a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-26 15:05:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-cinder-puppet-lint http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-lint/116ea24/ : SUCCESS in 7m 16s\n- gate-puppet-cinder-puppet-syntax-3-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-syntax-3-centos-7/5fb9302/ : SUCCESS in 4m 30s\n- gate-puppet-cinder-puppet-syntax-4-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-syntax-4-centos-7/ad122aa/ : SUCCESS in 7m 08s\n- gate-puppet-cinder-puppet-unit-3.6-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-3.6-centos-7/324e580/ : SUCCESS in 13m 03s\n- gate-puppet-cinder-puppet-unit-3.8-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-3.8-centos-7/024b953/ : SUCCESS in 12m 41s\n- gate-puppet-cinder-puppet-unit-4.5-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-4.5-centos-7/d39e166/ : SUCCESS in 12m 31s\n- gate-puppet-cinder-puppet-unit-latest-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-latest-centos-7/97c186f/ : SUCCESS in 14m 06s (non-voting)\n- gate-puppet-cinder-puppet-beaker-rspec-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-beaker-rspec-centos-7/9db40a4/ : SUCCESS in 17m 29s\n- gate-puppet-cinder-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-beaker-rspec-ubuntu-xenial/350d9e2/ : SUCCESS in 10m 21s\n- gate-puppet-cinder-releasenotes http://docs-draft.openstack.org/82/376482/1/check/gate-puppet-cinder-releasenotes/557a44a//releasenotes/build/html/ : SUCCESS in 2m 03s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/393b717/ : SUCCESS in 25m 02s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/10315e6/ : SUCCESS in 32m 56s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/c540070/ : FAILURE in 1h 03m 56s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/aed2d82/ : SUCCESS in 43m 54s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/82/376482/1/check/gate-tripleo-ci-centos-7-nonha-multinode/3139621/ : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"14ad00b7e4deb0b8e5ff4d9eb50602bbd6fcc78b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-26 15:10:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5cae091c3f0a23d86990c1f134539095553e545","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-26 16:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-cinder-puppet-lint http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-lint/2ec5c67/ : SUCCESS in 3m 04s\n- gate-puppet-cinder-puppet-syntax-3-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-syntax-3-centos-7/3860276/ : SUCCESS in 6m 24s\n- gate-puppet-cinder-puppet-syntax-4-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-syntax-4-centos-7/fc804dc/ : SUCCESS in 5m 57s\n- gate-puppet-cinder-puppet-unit-3.6-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-3.6-centos-7/207c185/ : SUCCESS in 18m 11s\n- gate-puppet-cinder-puppet-unit-3.8-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-3.8-centos-7/6a7a79b/ : SUCCESS in 13m 29s\n- gate-puppet-cinder-puppet-unit-4.5-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-4.5-centos-7/f3d00a1/ : SUCCESS in 13m 40s\n- gate-puppet-cinder-puppet-unit-latest-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-unit-latest-centos-7/ba8b8a6/ : SUCCESS in 14m 49s (non-voting)\n- gate-puppet-cinder-puppet-beaker-rspec-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-beaker-rspec-centos-7/bd26db4/ : SUCCESS in 11m 58s\n- gate-puppet-cinder-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/82/376482/1/check/gate-puppet-cinder-puppet-beaker-rspec-ubuntu-xenial/3ddbb24/ : SUCCESS in 7m 53s\n- gate-puppet-cinder-releasenotes http://docs-draft.openstack.org/82/376482/1/check/gate-puppet-cinder-releasenotes/e9938f2//releasenotes/build/html/ : SUCCESS in 1m 22s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/450d941/ : SUCCESS in 21m 40s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/ff34333/ : SUCCESS in 38m 42s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/68023e3/ : SUCCESS in 27m 29s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/82/376482/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/ac5a20e/ : SUCCESS in 35m 29s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/82/376482/1/check/gate-tripleo-ci-centos-7-nonha-multinode/2118482/ : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1602086c77e14a23a8963dc49cb3fb8083ea161b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2016-09-26 16:40:53.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70f1c952056d01797f007dbe71bb61c2657378ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-26 16:41:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"2254e91481d1c9336b24a4b370401a663fccf70a","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-09-26 19:40:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/5131/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2975/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/5068/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a40540b18e6d7837c7d3d539dbd16ee3aa327cf5","date":"2016-09-27 00:47:15.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"2d77becc972f8202111eb55bd1c3e9bff7a3519f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-27 00:47:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-cinder-puppet-lint http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-lint/8f65e86/ : SUCCESS in 2m 36s\n- gate-puppet-cinder-puppet-syntax-3-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-syntax-3-centos-7/cb17190/ : SUCCESS in 3m 00s\n- gate-puppet-cinder-puppet-syntax-4-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-syntax-4-centos-7/27229ab/ : SUCCESS in 2m 44s\n- gate-puppet-cinder-puppet-unit-3.6-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-unit-3.6-centos-7/89f50ae/ : SUCCESS in 12m 42s\n- gate-puppet-cinder-puppet-unit-3.8-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-unit-3.8-centos-7/3a41a7a/ : SUCCESS in 12m 23s\n- gate-puppet-cinder-puppet-unit-4.5-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-unit-4.5-centos-7/08a50bf/ : SUCCESS in 17m 13s\n- gate-puppet-cinder-puppet-beaker-rspec-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-beaker-rspec-centos-7/244075a/ : SUCCESS in 8m 31s\n- gate-puppet-cinder-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/82/376482/1/gate/gate-puppet-cinder-puppet-beaker-rspec-ubuntu-xenial/e5b95ef/ : SUCCESS in 10m 54s\n- gate-puppet-cinder-releasenotes http://docs-draft.openstack.org/82/376482/1/gate/gate-puppet-cinder-releasenotes/7d5c4c3//releasenotes/build/html/ : SUCCESS in 1m 00s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/82/376482/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/a797d98/ : SUCCESS in 28m 38s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/006b69a/ : SUCCESS in 24m 20s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/82/376482/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/9028609/ : SUCCESS in 35m 25s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/82/376482/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/c102ac9/ : SUCCESS in 24m 38s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/82/376482/1/gate/gate-tripleo-ci-centos-7-nonha-multinode/1b29004/ : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d002e31ffaefa450727bce6eb289e4b72b119b5a","revisions":{"d002e31ffaefa450727bce6eb289e4b72b119b5a":{"kind":"REWORK","_number":1,"created":"2016-09-26 13:32:09.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/82/376482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/82/376482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/82/376482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/82/376482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/82/376482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/82/376482/1"}}},"commit":{"parents":[{"commit":"2cad97c97a54ea9e88e1ca420a6372b2a66326f2","subject":"Merge \"Fix cinder_type to work with no properties param\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/2cad97c97a54ea9e88e1ca420a6372b2a66326f2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2016-09-26 13:31:18.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2016-09-26 13:32:08.000000000","tz":-360},"subject":"Puppet OpenStack Newton RC2","message":"Puppet OpenStack Newton RC2\n\n9.4.0 will be our RC2.\n\nChange-Id: I36db18a5a3bbcbfc6a80b9ec4dd1c772fd93cffa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d002e31ffaefa450727bce6eb289e4b72b119b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d002e31ffaefa450727bce6eb289e4b72b119b5a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2cad97c97a54ea9e88e1ca420a6372b2a66326f2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
