)]}'
{"id":"openstack%2Fpuppet-cinder~716961","triplet_id":"openstack%2Fpuppet-cinder~master~I35016b7eb2a314e713bfc5d3d9cb17360b2bad4f","project":"openstack/puppet-cinder","branch":"master","hashtags":[],"change_id":"I35016b7eb2a314e713bfc5d3d9cb17360b2bad4f","subject":"Make tempest scope more strict","status":"MERGED","created":"2020-04-02 11:49:25.000000000","updated":"2020-04-03 06:58:06.000000000","submitted":"2020-04-03 06:58:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716961-1585897086089-65716cc9","meta_rev_id":"37fa6ff01abd6d538daf5d1550d832327a739b9d","_number":716961,"virtual_id_number":716961,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 06:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-03 05:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-02 16:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-03 05:09:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 16:36:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-03 05:09:52.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-03 06:58:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bd6ada0bf502a4944d70fc904fe0819acfaf14e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cae9cc268830f34dd80a60a6972c2144397d6d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:48:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54aaa951163e4323ab85b94899e49335 : SUCCESS in 6m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8da4680e955c4e1cb6ac03fb4ac1bccd : SUCCESS in 7m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26f9714f6af74411ba79e27705f2ea10 : SUCCESS in 7m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8469a32f917c4332ae061a30ba9b4bce : SUCCESS in 18m 47s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7108d21265894eb5a99eaa9ebf19102c : SUCCESS in 19m 56s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72c62368f6b146abbd01071c6eb38f42 : SUCCESS in 16m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/743c31977e5b433d85ce333f3a207a89 : SUCCESS in 15m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e0c415001d8544918350abe085f1ee96 : SUCCESS in 48m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/761d5333dde143c19ddad24e10cf12f4 : SUCCESS in 49m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a157a6b68b304e958c5efad4a60cfd20 : SUCCESS in 38m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/da88d2073415452296cd9676b01bcb0a : SUCCESS in 36m 16s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/82ada1ea26c54a5192a5a91e3b4fd696 : SUCCESS in 59m 01s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f7ef047384a44b3a1268fee701dbd1b : SUCCESS in 1h 06m 02s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/79c2dec7702a4b839449c6c0b83c226b : SUCCESS in 59m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7f37a91336294e11af3ce3770d76d786 : SUCCESS in 46m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ee9b4c20118249ecb02bc4152938d530 : SUCCESS in 20m 24s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b63c25f92544129ad6c475ab6ce10c8 : FAILURE in 7m 09s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f06f906c44b4989a9e07e8dadba0bf7 : SUCCESS in 1h 51m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a2257de609c3736eac16669bb32a68ab34581d11","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-02 16:36:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a112b6d0fef7e9ce2fe72c97ebea731c8c078505","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-03 05:09:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ea3bf850e86bd9df2992840f4646356e73af63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 05:10:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b6972c8115fa8df2a3b6c26595956fd567a3947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 06:58:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a2d838fab724855816a2c757826f2e8 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0ff0f056f5f4a598dfbc6117af18841 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b86848c740454093b291ac666fb37efd : SUCCESS in 4m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bd2a11a663844bb388ef8a06288b0242 : SUCCESS in 17m 49s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/0e2ae24ffbd04f63a975b35b9f50a1bb : SUCCESS in 17m 25s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b82c53e61d034cfd9ec6d48299382a54 : SUCCESS in 16m 05s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4afe7655229c42c9adcf9b313f9c142f : SUCCESS in 53m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/018f451275a346d1ab80d9ede208471f : SUCCESS in 1h 01m 18s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3dada7f7ba4241778abb5dd5f8429e90 : SUCCESS in 1h 00m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/931184a86b7a4bcbbca5dae85a11c154 : SUCCESS in 45m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5e641dc41d0f41dba100a7b6efb6bc0f : SUCCESS in 19m 43s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/65153a7f127d467d943121fb7402eaf0 : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"37fa6ff01abd6d538daf5d1550d832327a739b9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 06:58:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0973f285979b05414e0f411af95e5970e447311d","revisions":{"0973f285979b05414e0f411af95e5970e447311d":{"kind":"REWORK","_number":1,"created":"2020-04-02 11:49:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/716961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/61/716961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/61/716961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/61/716961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/61/716961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/61/716961/1"}}},"commit":{"parents":[{"commit":"8af028037410d9847a44629ddcda9d9861188b6f","subject":"Merge \"Remove Dell EMC PS Series Driver aka Eqlx\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8af028037410d9847a44629ddcda9d9861188b6f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:49:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:49:11.000000000","tz":540},"subject":"Make tempest scope more strict","message":"Make tempest scope more strict\n\n... to prevent tripleo gate job from getting affected by newly\nadded jobs.\n\nChange-Id: I35016b7eb2a314e713bfc5d3d9cb17360b2bad4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0973f285979b05414e0f411af95e5970e447311d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0973f285979b05414e0f411af95e5970e447311d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
