)]}'
{"id":"openstack%2Fpuppet-cinder~719914","triplet_id":"openstack%2Fpuppet-cinder~master~Idaa1c767e8cab645567b920dc9d377abf509ec54","project":"openstack/puppet-cinder","branch":"master","topic":"provider-credentials","hashtags":[],"change_id":"Idaa1c767e8cab645567b920dc9d377abf509ec54","subject":"Refer keystone_authtoken to find credentials","status":"MERGED","created":"2020-04-14 11:10:48.000000000","updated":"2020-04-18 09:19:21.000000000","submitted":"2020-04-18 09:19:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719914-1587201561623-50745a22","meta_rev_id":"95ad1a41d8c20ec7fa40a24841e8443e8046b330","_number":719914,"virtual_id_number":719914,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-18 05:41:35.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-18 09:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-17 18:56:58.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-16 08:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-17 18:56:58.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 08:22:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-17 18:56:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-18 09:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"558f4b2ede48ee2c8282c681da3e700308c1a221","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-14 11:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbb05de2a7afad54e1ed9cc5f9dd2b79452f509c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-14 11:18:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e615c594c4eb3c96e1e60be5863de59429ec10a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-14 11:21:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f613191c47e7e44e062784234fada82f95370286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:28:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba0305117bae4105b41d88e817aba9e5 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb1d90525be84d98a2110feb4767e47d : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d9235d8f92549c79b0b5bf79444d5a6 : SUCCESS in 4m 41s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/79b009b49fe74dd889f5b7db57e98752 : SUCCESS in 19m 32s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2cee6fb18a78469fa7b0d359b7bb278f : SUCCESS in 19m 07s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef38cdbfa23245fba87de49149c841d8 : SUCCESS in 13m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/009e2bccb16f490e95cfe6fde3f50670 : SUCCESS in 14m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f82755554eda4f95a0572b2b1d465a49 : SUCCESS in 1h 14m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47954f5bb68041f3b3e188dd0cc578b2 : SUCCESS in 51m 39s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f79e63a6329346b9b318c2c04552b096 : SUCCESS in 49m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9bb16d227fa84d6896a2c26d6a5b24d2 : SUCCESS in 40m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42824a2a75054447a0ed15ebe568d786 : SUCCESS in 54m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/20c76d4eddca4a9ca78cb58ead6e14e1 : SUCCESS in 1h 04m 48s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cee9e2c7de0b4a818aa1cede0f9aa05b : SUCCESS in 1h 04m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/79c00031475d48ddb145da21f19fab5f : SUCCESS in 47m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ad8e1baffc924436a732c66053ad0203 : SUCCESS in 18m 41s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/63f1ea29fb11402086bcd6ed154bf87c : FAILURE in 8m 21s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9f453abdf94426b8c88f3b3c66ca3e0 : FAILURE in 6m 54s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a750cb0f2e7b4e398e462760474e6af3 : FAILURE in 1h 50m 45s","accounts_in_message":[],"_revision_number":3},{"id":"f86c5068ac63e4d61609bde151677345cfb9e32d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-14 13:29:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4133d1dc6a72c468b39b878ff7045edcac23fafc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 15:35:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22941c7216e945f48df4e863613a5cc4 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a598857c759e432e8a4d6f430229fcfa : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9906504c4c804ad4b03718f2d45178de : SUCCESS in 5m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3b18ea72177e4790b4c8f941ab7faa81 : SUCCESS in 19m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/033eeb24feaf4e1591ec4c19d0b1abf3 : SUCCESS in 20m 01s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8557ec3a9f2c4b47aa55c7e6843a9e75 : SUCCESS in 15m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61109e1834574919987fb633bf11e295 : SUCCESS in 21m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f13c333877374e97838af6e4fa667553 : SUCCESS in 56m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6e3230173f742c492bb820adbda98ae : SUCCESS in 58m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/048de630b26a4b3d8b92b03e304a86cf : SUCCESS in 40m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/40f64b41393e4e0ea089bd48e7126fd0 : SUCCESS in 36m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c4e52ed99af243dcb14ed3d42bd6ed04 : SUCCESS in 53m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/811e7739fd744f51ae7a4b8bce20e802 : SUCCESS in 1h 11m 40s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/29430002dcd143e19ff6c2618c8b4808 : SUCCESS in 1h 06m 17s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d1c4042339814e99a6bf25af33bcb3ba : SUCCESS in 42m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b90a2adfc4da4a87b7d005718d8a1a58 : SUCCESS in 21m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/cc07e65ebf424951ae0784af9444bba1 : FAILURE in 11m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb7fff4c85bc46509d47b3d10d89f2e3 : FAILURE in 5m 43s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7f6fd82a6e1e4b3196f8a67485c9784d : FAILURE in 1h 48m 49s","accounts_in_message":[],"_revision_number":3},{"id":"78a65902618a0dffcdf974ac4e3c9f9bb4d34438","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-14 16:19:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dfdd0523f9782444602a789371da2481b99bd0c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 18:10:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/016be998de984b9bb92c0bb276457a83 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddd0fa57ce574a7c80d8ba88d73a224c : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a2bd82a703c4996a9d7737801a73416 : SUCCESS in 8m 11s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/afed91cd61c04eee91d1a27376a441be : SUCCESS in 18m 26s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1b72d47cb19e494dafc14f6fa8957e04 : SUCCESS in 18m 19s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30afd4a2f2b946c59ad754932bef79b1 : SUCCESS in 13m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36adf1d5fabf4f9bb3bbf6808ba49f34 : SUCCESS in 14m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/059d41d477df4b6dbc18164cb0808946 : SUCCESS in 47m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa107dcd43654c32a7141cf8737098fd : SUCCESS in 51m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a68c8252e9945cdaeb94a0744cfc35b : SUCCESS in 55m 43s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/034d16a23a6e45728be61dff95f7cea3 : SUCCESS in 36m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c1b619257f07410584cf016eef573e9b : SUCCESS in 54m 00s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/43e65d5824cb4ecf95aa2ebd9ea1a8af : SUCCESS in 1h 05m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a56ca19f35704ebb8582e59d6b697603 : SUCCESS in 1h 14m 00s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ba3949ddac3f46ceae05a6c0c05fbc4a : SUCCESS in 44m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fdbad082e12f4bcb9826f42cdfcdd0d6 : SUCCESS in 16m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/463e8dcb00e1469f958fcaddc9a00892 : FAILURE in 8m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28e2939e1fc04c1db5ba8ba6e4cd83bd : FAILURE in 10m 26s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/05af0a875221407eb210f31fabebf59a : FAILURE in 1h 43m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c8c85592dd748efc7a9fc8b2ce53b6328c6942c9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-15 02:16:45.000000000","message":"Topic changed from os_region_name to ks-region","accounts_in_message":[],"_revision_number":3},{"id":"ac128aac343b7308878e5a98e2cf65f646bb7ff9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-15 02:17:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2e3251dbd62611b1bcc282f648b5e3f5ba2c20b8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-15 02:19:43.000000000","message":"Topic changed from os_region_name to provider-credentials","accounts_in_message":[],"_revision_number":4},{"id":"25cf9d86cbe9bd0d84f8b4ab3b8052040a7385a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 04:28:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47c21b5729c640da972dc984e795258d : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8d8912ac6a94fc29609f99f939ad9f0 : SUCCESS in 6m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/076451242b7447eb8f75afb340ced460 : SUCCESS in 4m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/58bdad957cd24ae1985c97b105775a08 : SUCCESS in 15m 58s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f4799d614e294dbdab67181116632be4 : SUCCESS in 13m 21s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/857e1e0a5ef74606989fbe3497d169de : SUCCESS in 13m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f7bb84283324254a47839c8ebb942f9 : SUCCESS in 13m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e46bd2d867cc4979ab537213c3503b9a : SUCCESS in 55m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a209775a330469d8056bbcc7ca52feb : SUCCESS in 48m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5f6e40da04f4266be912e2fd1c6642c : SUCCESS in 47m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e4fa2b70565246a6897edf7d0cd4b84a : SUCCESS in 32m 58s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/464908e6e86040afb8697907c2b36fcf : SUCCESS in 1h 00m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/10ff220dd8a24f6b80528be87807276f : FAILURE in 1h 27m 35s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/554cfd0bd26644aa8268209f81afa26c : SUCCESS in 1h 09m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/40ee41405f3c46518b588f4d79ea22d5 : SUCCESS in 46m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/aeaed7b81aeb44a5a2a2a024b4215483 : SUCCESS in 19m 41s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7e10da3dc0664b9ba4912bf5c89f203d : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9c3810a5628435b9aa8f64d8cf84b75 : FAILURE in 7m 18s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/633bb746c2994548acb5f9c6aa77cc6e : FAILURE in 2h 09m 20s","accounts_in_message":[],"_revision_number":4},{"id":"54a2d99804b1ce9e14ae7c08edd448e76b9fc269","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-15 05:24:18.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"87e3d434990f2e050164ee8f8010e68257ed6f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 07:20:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e3a4706b6704438886c8f33f5eec6fb : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32e2e0d0e4af4ba9a121004d072126b7 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ec1c8502e97496980cffddaccdf1a9a : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0605cd46259f4f8dabcce35d4ee8945c : SUCCESS in 18m 00s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c9e213475c084843ace6b592cf96448c : SUCCESS in 18m 05s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5db782295d14fafb9f475ccf7b2d3f3 : SUCCESS in 12m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d375ee52b48f4993a4ae54f98abad697 : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/448b86f380744e4ab5c46590d2dd60a7 : SUCCESS in 48m 19s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8069ff805dd3461cb3f51f5d19a83f96 : SUCCESS in 50m 42s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82252a047cf64fff82aa857261a0086c : SUCCESS in 41m 17s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/46e43adde5a042c5a23960dc502db0a5 : SUCCESS in 37m 54s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d5787aaf66004a428278f777ed923734 : SUCCESS in 1h 06m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42eb8f19fee44c249b0e6780767ea8ac : SUCCESS in 1h 00m 00s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/423bb877a8014c5fb4cc94589c513ac9 : SUCCESS in 1h 07m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/20994624e40f464e83f781621aa438ce : FAILURE in 48m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/18bc33adf44642ec862aedbb9d40082e : SUCCESS in 20m 34s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6fcb30d637cb4eaba03dfbeb73d3efbe : FAILURE in 7m 44s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0582ba75add94f84bc5316ed1b3d751d : FAILURE in 7m 12s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/88a54d5f9f294a129d345a17527fbc2e : FAILURE in 1h 55m 09s","accounts_in_message":[],"_revision_number":4},{"id":"1c46140f296175c36ae8592a0f5a621d28f36bbd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-15 22:47:51.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7d41945d9df82579b21f9ff5d8848519983e9e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 00:41:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c0a067813684cccbced242e3c838cb9 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f39262e09d7e4b31a2f6d9bdbae0f890 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c15833ae6d554cf8aa901e8e5191ee82 : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4bc33e0ceba540209161b2ef63bd3486 : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a77536c8b0e247f58d47f71056b84ec6 : SUCCESS in 14m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bac2fc2c9b714a648e1b951358fdc063 : SUCCESS in 12m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bade45eacb4745ba9120c7f1106b379c : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b3e2c2cc3bd1475dbd0923813b33794b : SUCCESS in 47m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81f5f9be915047f1bbe4ce193b453929 : SUCCESS in 50m 47s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/902857faa4574d9080fc6dac998b3267 : SUCCESS in 41m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6a7c246e2ffb496e8847c5faccbe0897 : SUCCESS in 36m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/535ce80d9cb64f4eb5aeab11dc56a8a9 : SUCCESS in 54m 52s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bf3bad0428bf45ce91f1a33e959133f3 : SUCCESS in 1h 07m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fef5dc2cda794186a9f0ad14ba571fa6 : SUCCESS in 1h 04m 18s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d094d3acda514c9f8eba9209aefe1425 : SUCCESS in 42m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/320fab342a3749fcb9a2b7be576ae233 : SUCCESS in 19m 56s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d4d42f79618748bf8183a46f1a9926c4 : FAILURE in 8m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b79be191a63c43c4b1fae7676285143a : FAILURE in 6m 39s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7ed3488693a74241bd6c7aae74672d3e : SUCCESS in 1h 48m 24s","accounts_in_message":[],"_revision_number":4},{"id":"61fc93431a5bb0275d1fdc02e74108ef39f6b4fd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-16 08:22:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9082324189819c4eb6fca67abe3065d49b782cd4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-17 18:56:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0df09c1bffcf458334fb98d907a1f0d727346e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 18:57:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cba82332f1e5bf4846e1c6e06f77a9a472dcb099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 21:11:34.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af36c2825d504248962bba10a0e6cf32 : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23afcd9f109f481598f06881e065d067 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e03fa0cff30842258f24725cc7147000 : SUCCESS in 4m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d38c1107fb684a57ae2780a89156fd35 : FAILURE in 7m 43s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/05606d1abac64c7ab4459947246f4f15 : SUCCESS in 21m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5dcd2fddbd1499281cc4963f4df1c1a : SUCCESS in 15m 39s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/194b2a80fef14d90b8d928dd7962b083 : SUCCESS in 1h 18m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce3243df07004900a4da8b0fb6e97fd7 : SUCCESS in 1h 22m 02s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/81d5040d34b14a60bc9a080dbae94243 : SUCCESS in 1h 02m 33s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c6bb02b01b774b4cb9e7148c7632647e : SUCCESS in 47m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ee86a6fbe1774cc9a88807c748b65f16 : SUCCESS in 18m 28s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a2dc06d81cc145bfa5fdabac2a40341a : SUCCESS in 2h 02m 16s","accounts_in_message":[],"_revision_number":4},{"id":"208538e7f6e9aecff1b2db8c93c876df5f7edca4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-18 01:22:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a797babe8e61770bfbf4f94c878da9a12508b55b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 03:11:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9ea1340c5784cc09226205dc512cc35 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/729a5ce95e7344cbb23017789dc5ae2c : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57149ebc661f4621a4647faa695c1ae5 : SUCCESS in 4m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c5dfe023064f490bbf1c4eba2bf9064b : SUCCESS in 17m 32s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f1f728c8bab247a9a9be1f4aef82c048 : SUCCESS in 17m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d46c3834fe2042cea30b2633d97c79b0 : SUCCESS in 14m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2cfb5e8bda94a7da25d164c279a26a0 : SUCCESS in 14m 07s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/747fbd563146463e85c5a25ab08d8756 : SUCCESS in 47m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efaeaae66860499f858ffafed0b19ff4 : SUCCESS in 51m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7140109bdd3402f969af8f650c2f811 : SUCCESS in 39m 57s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0cfed6b54f594b5b9becb2c8ff6016a1 : SUCCESS in 35m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c87935093a904cb1b02ab6f11da0587b : SUCCESS in 56m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aed62dc107bc4a979709175c0a3d3a1d : SUCCESS in 1h 02m 47s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b167d781d2a45539bb939db7a8a510a : SUCCESS in 1h 06m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f26ec0514984a3d81e0f16cc9f627e8 : SUCCESS in 41m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/da000ede39934eba9b6773dd1db5fa94 : SUCCESS in 19m 57s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/37ba0e3398784c678ccf4f3929a4d2c2 : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d5a1d0c6df4416dbb1295127384eae3 : FAILURE in 6m 17s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93409f7365e14e14b310968400429afc : SUCCESS in 1h 44m 04s","accounts_in_message":[],"_revision_number":4},{"id":"1c5c85e0888fc25634858f1aa890a469eb4e8ab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 03:11:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"62fc7dcaebcd9ac697159c895783deb3e42d3a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 04:51:12.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f46b46ecf8f42f0982f7ac79ccbe953 : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e5c711ad4cf4213aa06f18ae67633ff : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/616cccddfb6449a5a9eb6114d5374bfa : SUCCESS in 5m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c0ec84d6dee5428fa2c7ddca8253c545 : SUCCESS in 19m 54s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7dcf64b8666e4f50b1b8ac8ac1fbc4a5 : SUCCESS in 18m 59s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d78aae872188434ca1ad1708e676d74d : SUCCESS in 13m 43s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d15e4eec2b254f92b9f858f3faec003c : FAILURE in 30m 52s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c3f01bf52b44b269214fc7dbf0a491d : FAILURE in 31m 53s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ff35433dc4164ec1b000259b1b066682 : FAILURE in 32m 40s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d5967dac2e4647a7965e0e31aad2ceb5 : FAILURE in 29m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0f0561bf632d495f81f57c248fd034d4 : SUCCESS in 25m 26s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/079e0ccec4984412a6167c9998a29175 : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":4},{"id":"56f711789f2f9f2110466219a066386a726b6f15","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-18 05:41:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5da5188bb6e7db124bd252eefbbccedcef1f0dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 07:26:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63b803fec6a54eeea3cf7b6009b80193 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cef2c4e2d2a480182023323ff140e05 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/caa8c87da7fc4ddc99c935782c0f2551 : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/466b78f29aed4487b38a792ec124a346 : SUCCESS in 19m 33s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/57dbea5302ff4e4b947948dbe367b237 : SUCCESS in 14m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff3be4be11e44d02ba232ae9b56d60a2 : SUCCESS in 14m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9db798fa0603484b95f634b76d05ed9b : SUCCESS in 16m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bbab52df87794de5943a6d96ec5faf51 : SUCCESS in 46m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ac18b9bfd0848ac8742cc64bc909558 : SUCCESS in 52m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f178808b2554588b24567bc3b3eddc1 : SUCCESS in 1h 03m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5120c20ee14d456fb38bba2a51117902 : SUCCESS in 56m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc7efc8c934945c78cc180080cbc0146 : SUCCESS in 54m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7593aba50c464b6095b43525bfc0f037 : SUCCESS in 1h 20m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8112de7ba59f4350bb8530abe2f3508a : SUCCESS in 1h 04m 37s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/381b0ad442224022b74559179bb9dda7 : SUCCESS in 1h 03m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a2e5bb0082b14c65a1fc129bac1a09d9 : SUCCESS in 31m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/cbdfabe56b994d959e8ceb62d874ce33 : FAILURE in 9m 10s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcd923fb0bc14ce08a2374651647359b : FAILURE in 9m 02s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a505ac20ba3548008f4dfd8b030e271e : SUCCESS in 1h 43m 44s","accounts_in_message":[],"_revision_number":4},{"id":"1c32574bc84b81d874f065a00e25e163ea5f57d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 07:26:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"862498c2cb6fcfd338f0d23cfe416158af435766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 09:19:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b69ae1b504b425398503a49acb884e8 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb3d8f9fd5cf455eae7d0aa67f6e2fac : SUCCESS in 5m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db24bd4265164257b34918f269403701 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/076ef94a22504d8d9e2d40eda56937f1 : SUCCESS in 20m 26s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/bedcf168d142488bb2c5fe4e6c6b4e1a : SUCCESS in 18m 30s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdbdaaaa541e4d8db0ef4774ec8f5e2d : SUCCESS in 16m 46s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bad6ee76280f4cad918609b55e069bab : SUCCESS in 1h 05m 16s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ebef762045f44fdb90ca543102bd4122 : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/16f383c429a041f1971a19d84f391e1e : SUCCESS in 1h 14m 53s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/544764a3c1d34eec968ce3f8c1140c7a : SUCCESS in 44m 43s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/769ac96e698e45358a3b4915a53b248f : SUCCESS in 18m 45s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4631413c18854c9a8a0564fd84b08301 : SUCCESS in 1h 50m 24s","accounts_in_message":[],"_revision_number":4},{"id":"95ad1a41d8c20ec7fa40a24841e8443e8046b330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 09:19:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"29e221fe55236031d9d47d62ca9588b391aae6da","revisions":{"72fa544820effeb3dc1402a92289f4fda275ec77":{"kind":"REWORK","_number":1,"created":"2020-04-14 11:10:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/719914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/14/719914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/1"}}},"commit":{"parents":[{"commit":"3b612bba47ca972069950e83351f9e2820e6ae3c","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3b612bba47ca972069950e83351f9e2820e6ae3c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 11:09:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 11:09:23.000000000","tz":540},"subject":"Use keystone_authtoken/region_name to detect region","message":"Use keystone_authtoken/region_name to detect region\n\n... because api/os_region_name is not set since [1].\n\n[1] c2d618cae70ba1be7dd8496ce7665ac3abaf8775\n\nChange-Id: Idaa1c767e8cab645567b920dc9d377abf509ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/72fa544820effeb3dc1402a92289f4fda275ec77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/72fa544820effeb3dc1402a92289f4fda275ec77"}]},"branch":"refs/heads/master"},"e752cc8200cd9594f0cde29f83d91094b0c53863":{"kind":"REWORK","_number":2,"created":"2020-04-14 11:18:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/719914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/14/719914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/2"}}},"commit":{"parents":[{"commit":"3b612bba47ca972069950e83351f9e2820e6ae3c","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3b612bba47ca972069950e83351f9e2820e6ae3c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 11:09:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 11:18:38.000000000","tz":540},"subject":"Use keystone_authtoken/region_name to detect region","message":"Use keystone_authtoken/region_name to detect region\n\n... because api/os_region_name is not set since [1].\n\n[1] c2d618cae70ba1be7dd8496ce7665ac3abaf8775\n\nChange-Id: Idaa1c767e8cab645567b920dc9d377abf509ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/e752cc8200cd9594f0cde29f83d91094b0c53863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/e752cc8200cd9594f0cde29f83d91094b0c53863"}]},"branch":"refs/heads/master"},"df76bcd2e4a80c44ad295257bfe8594e863f1a33":{"kind":"REWORK","_number":3,"created":"2020-04-14 11:21:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/719914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/14/719914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/3"}}},"commit":{"parents":[{"commit":"3b612bba47ca972069950e83351f9e2820e6ae3c","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3b612bba47ca972069950e83351f9e2820e6ae3c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 11:09:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 11:20:47.000000000","tz":540},"subject":"Refer keystone_authtoken to find credentials","message":"Refer keystone_authtoken to find credentials\n\n... because api/os_region_name is not set since [1].\nThis patch also fixes incorrect lookup about domain parameters.\n\n[1] c2d618cae70ba1be7dd8496ce7665ac3abaf8775\n\nChange-Id: Idaa1c767e8cab645567b920dc9d377abf509ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/df76bcd2e4a80c44ad295257bfe8594e863f1a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/df76bcd2e4a80c44ad295257bfe8594e863f1a33"}]},"branch":"refs/heads/master"},"29e221fe55236031d9d47d62ca9588b391aae6da":{"kind":"REWORK","_number":4,"created":"2020-04-15 02:17:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/719914/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/14/719914/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/14/719914/4"}}},"commit":{"parents":[{"commit":"3b612bba47ca972069950e83351f9e2820e6ae3c","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3b612bba47ca972069950e83351f9e2820e6ae3c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 11:09:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-15 02:16:34.000000000","tz":540},"subject":"Refer keystone_authtoken to find credentials","message":"Refer keystone_authtoken to find credentials\n\n... because api/os_region_name is not set since [1].\nThis patch also fixes incorrect lookup about domain parameters.\n\nThis patch also removes useless and incorrect handling about keystone\nversion, so that domain parameters are correctly set.\n\n[1] c2d618cae70ba1be7dd8496ce7665ac3abaf8775\n\nChange-Id: Idaa1c767e8cab645567b920dc9d377abf509ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/29e221fe55236031d9d47d62ca9588b391aae6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/29e221fe55236031d9d47d62ca9588b391aae6da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
