)]}'
{"id":"openstack%2Fpuppet-cinder~721123","triplet_id":"openstack%2Fpuppet-cinder~stable%2Ftrain~I5dbc65df8476a3f69876c3ca14bb2500ea25134e","project":"openstack/puppet-cinder","branch":"stable/train","hashtags":[],"change_id":"I5dbc65df8476a3f69876c3ca14bb2500ea25134e","subject":"Deprecate cinder::api::os_region_name","status":"MERGED","created":"2020-04-19 23:17:20.000000000","updated":"2020-04-20 21:00:17.000000000","submitted":"2020-04-20 20:58:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721123-1587416325226-354c625b","meta_rev_id":"b7ec9c150e62bc088d9fe7a49fd497eb132d6701","_number":721123,"virtual_id_number":721123,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 20:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-20 19:30:22.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-04-20 14:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-20 19:30:22.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 14:43:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-20 19:30:22.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-20 20:58:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10e93de88f51086137d74ba92c62260cc0028766","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 23:17:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f299075db443053d0b83a0f93a2a1270d70bf3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 01:04:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5ee6fc0dfd347b9a2ceaf32b618a629 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dc4e7bab8fa485b81e0f4baef778f43 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca5dfe820d704dd7a9714aa656754f99 : SUCCESS in 4m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d5f8887f746444f4bec6e6e6dce5c7ea : SUCCESS in 23m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b23aae230b3549a99fdfaa4699efb6a0 : SUCCESS in 17m 15s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/b65cd027730b41bfa965dab071344e97 : SUCCESS in 17m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a894f2ffe73d4ec2bdca49f8a0f93f38 : SUCCESS in 16m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a88ce1ba85594dae9dafdc0ac46f0af0 : SUCCESS in 12m 25s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f229018620364abea5af50b4b215a290 : SUCCESS in 50m 49s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5ed09e2a0b854a4c96642304847645ae : SUCCESS in 1h 07m 39s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cd84e823c4f7452d8e050238274614d5 : SUCCESS in 1h 08m 03s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1cf8689a06ef4136b07fdcbf66562eac : SUCCESS in 45m 05s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/65eb3b50d7ec4141a06d3db386893dd2 : SUCCESS in 42m 21s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4cc82e631314e46a6d9c902b9aab615 : SUCCESS in 49m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/349b12e284974b618faddd6cbfecbef2 : SUCCESS in 34m 42s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2beca0b73596449c8dc7382cd4abb044 : SUCCESS in 29m 44s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/74fd34d1adba43ec83fb57bb4a565485 : SUCCESS in 52m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6d77bfbb55704c36b518854125df1857 : SUCCESS in 1h 07m 56s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7df2f51b184b474c8dd066961ee659fe : SUCCESS in 1h 01m 46s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/387dcdd3a27c4816af196b04acce3d8f : SUCCESS in 38m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ec312decb79f4f25968ea8013f76cd5a : SUCCESS in 18m 56s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f7e96f49cac4fdf8a887cdabbbd1fdf : FAILURE in 15m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f8c31b83fe14836aa86d6cb72780540 : SUCCESS in 4m 36s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0c20f39fe7c846d9810ea78d5618c8f1 : SUCCESS in 1h 45m 55s","accounts_in_message":[],"_revision_number":1},{"id":"57f89138ffd78b451d7eae22b0ed93ebb27ca59b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-20 14:43:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"20c23120a86da6890264a3a138b428f0320bfea9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-20 19:30:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a38bc37a7f98267901d3051944519464fb2b499a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:30:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d104292c86ef34b9e012376daf79a6fda0169b8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:58:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e1a3e9f12794314b85ef9f5e6b26b72 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5a4d50819b947f4863a85944b0db240 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d984ea639e7045babb2a409155eeb3db : SUCCESS in 6m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/677c09e659904e9da7e9c42b285d3f9a : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02c9f75f703746099b6009230568ff0b : SUCCESS in 11m 32s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9efbb573182846b9a8978b0a7be7ce9e : SUCCESS in 52m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e73c364053264245ad9982033833dcd7 : SUCCESS in 59m 55s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8f06a476af4f4fd1b97208384e56efd6 : SUCCESS in 1h 04m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/59128f325c704c758d68f0bdafeb6c3c : SUCCESS in 46m 01s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fce8a12008af4439b96a16aaba0bf586 : SUCCESS in 25m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47e45377b607472b94c69c878712c35a : SUCCESS in 4m 07s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/915de9df7bc5432cb936b5d5771c4dff : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a34762240071258e03ab43ccf3fcbe63fb528cca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:58:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7ec9c150e62bc088d9fe7a49fd497eb132d6701","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 21:00:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/504dedcbf8824598bde0f793737a5c01 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54fcada665445b2f28a5d785e6cebacf08561935","revisions":{"54fcada665445b2f28a5d785e6cebacf08561935":{"kind":"REWORK","_number":1,"created":"2020-04-19 23:17:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/721123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/23/721123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/23/721123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/23/721123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/23/721123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/23/721123/1"}}},"commit":{"parents":[{"commit":"298d2ace77c151a58f3fb706f27d2018d346a5a2","subject":"Merge \"Make tempest scope more strict\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/298d2ace77c151a58f3fb706f27d2018d346a5a2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:06:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 23:17:08.000000000","tz":540},"subject":"Deprecate cinder::api::os_region_name","message":"Deprecate cinder::api::os_region_name\n\n... because the actual parameter, DEFAULT/os_region_name was deprecated\nin Pike[1], and removed in Rocky[2] release.\n\n[1] da93e11794d2a7479b037f96fc5aef83f98be8dd\n[2] 50e6b7af3662ab89eefc565c182ef14bb20f08d3\n\nChange-Id: I5dbc65df8476a3f69876c3ca14bb2500ea25134e\n(cherry picked from commit c2d618cae70ba1be7dd8496ce7665ac3abaf8775)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/54fcada665445b2f28a5d785e6cebacf08561935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/54fcada665445b2f28a5d785e6cebacf08561935"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
