)]}'
{"id":"openstack%2Fpuppet-cinder~742016","triplet_id":"openstack%2Fpuppet-cinder~master~I831cd341a838280bdd91bdca94af1a860cdc8006","project":"openstack/puppet-cinder","branch":"master","topic":"rebrand","hashtags":[],"change_id":"I831cd341a838280bdd91bdca94af1a860cdc8006","subject":"Support for PowerFlex cinder backend","status":"MERGED","created":"2020-07-20 19:39:57.000000000","updated":"2020-08-06 02:05:33.000000000","submitted":"2020-08-06 02:04:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742016-1596679468255-145a67d5","meta_rev_id":"9fcd16a9fa27ce09bf7460e9349b98c8c75c3122","_number":742016,"virtual_id_number":742016,"owner":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"date":"2020-08-05 22:39:43.000000000","_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 02:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-07-29 19:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":2,"date":"2020-07-27 22:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-08-02 23:31:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":1,"date":"2020-07-28 15:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2020-07-29 19:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":1,"date":"2020-08-06 00:16:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 20:36:35.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-28 15:53:54.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2020-07-29 19:40:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-29 19:43:33.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-08-02 23:31:48.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2020-08-06 00:16:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-06 02:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60660b4849eeea7c754d53fc690ce09a8a381424","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-20 19:39:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f55e9d8d60615c2c21aec40a62dff629a535eb8","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-20 20:00:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98b1f819fcca25359e423dac256451632723d658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:52:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5d2b9832b754ce293103ab49ee3c860 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b86dca385564c61aa6211b67b95be61 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ee33cb252ff4a299a8feaa1beef699f : SUCCESS in 5m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/46f0288dade24795bf93d22a30a0951a : SUCCESS in 18m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2c641ce118cc4c99a8791348a3522aa3 : SUCCESS in 13m 47s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/653a13ba22a346eb87421aac70fe8919 : SUCCESS in 14m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dacc3cde61546e9833405c68338d703 : FAILURE in 7m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2dc7efa37e354f29bb5c92ac9c9d0c49 : SUCCESS in 52m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1d3709c06944017b7eebb57f009ed34 : FAILURE in 39m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0500136414904812b35509bc074afbe9 : SUCCESS in 48m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/755a95dad97541ed971698dbecff3e05 : SUCCESS in 29m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/96d8c54270af4b26aaa120614e9e0854 : SUCCESS in 53m 13s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0250d571d6204b6d81796a5a68239e20 : SUCCESS in 1h 33m 50s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bf2369bfd5234f72aef5a79af3fb9615 : SUCCESS in 1h 32m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6d5a2d028efb441eb31e2e2ffb56d543 : SUCCESS in 1h 05m 30s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d57ae6a21979443681dd107c94b6701c : FAILURE in 8m 16s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9ecdeb78bbbd494c9fef1ae0c3268e0a : FAILURE in 8m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2437815a09ae4554a735bc3e4dd26f6b : FAILURE in 6m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/496a6bf55e0d40678e4a365d2e3b1b3f : SUCCESS in 3m 45s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c51ed2a371544c68921932ccad0d115 : SUCCESS in 1h 46m 15s","accounts_in_message":[],"_revision_number":2},{"id":"c246bd9be30d71531562055890ae8cadfb5b7c31","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-21 16:50:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b924c8cf4f18daad1a09356d0dbbfe4e21796e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 18:32:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c77053a74aba42f1afbe2cf273e99551 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec7c417b6e6c47c5a2b0ad8e48fb0699 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/153e013db654455dbcaffc50557f2ffa : SUCCESS in 5m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/55ca9cf33bed4147b73a971cf8f93bfa : SUCCESS in 17m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c94a01df12af44879bd974e05776ceea : SUCCESS in 18m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e59b775fdfc746eaafd9236d2f0feab2 : SUCCESS in 15m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2203124e67fb4b31a8f7b3be2084301b : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c4985d9c93a4407a8b3e05e61c6c05c9 : SUCCESS in 37m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/821828e121ae4029b10d171d0a3556b3 : FAILURE in 53m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06e8aaad64254b54b8e60040c57ac4bf : SUCCESS in 54m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5455d5ed97014cc394e4f1bfea10b9a9 : SUCCESS in 41m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ef71c16a8fb144fcaf7b36ebce266d2e : SUCCESS in 1h 33m 44s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0b237bde99a14a2b98d952e6c540052f : SUCCESS in 1h 35m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e736d7ed2a3942f798b82108971142b1 : SUCCESS in 1h 30m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/158cb30ca7e0419cbde01ad9442e1c71 : SUCCESS in 55m 34s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a5c94268ac6146c1b4613d1fe1c41b83 : FAILURE in 9m 24s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/960a66b4ddef47c39adaf4539d3e8e72 : FAILURE in 8m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34f0bdaa53c14155ba65dd8a7c11aeae : FAILURE in 7m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78011587d84a40109d2c13e17f261864 : SUCCESS in 5m 27s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f6a07c20c04840d4be705d18f74a5647 : SUCCESS in 1h 31m 37s","accounts_in_message":[],"_revision_number":2},{"id":"70d87a364eb2bf20bb4ed685a3d168aa52684913","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-22 15:35:36.000000000","message":"Patch Set 2:\n\nPlease set Depends-on against cinder patch[1] to make sure that this patch is merged after actual implementation is merged in cinder.\n [1] https://review.opendev.org/#/c/741158/\nI\u0027ll hold my review score until I have more detailed look about implementation later.\n\nFYI. currently beaker job is broken and we need [2] to unblock it.\n [2] https://review.opendev.org/#/c/742146/","accounts_in_message":[],"_revision_number":2},{"id":"b04804083df06c8c749ffe3d5c3927accea91ebe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-22 15:37:36.000000000","message":"Patch Set 2:\n\nSorry. I mentioned wrong cinder patch. I should be this one instead[1].\n [1] https://review.opendev.org/#/c/740522/","accounts_in_message":[],"_revision_number":2},{"id":"1c8c165372809d0ad8d2b79e4759745c82f7a46b","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-22 16:51:48.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b429d5e46e58e6e1176c62e79ae2665df571d3b6","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-22 16:52:17.000000000","message":"Patch Set 3:\n\n@Takashi update the Depends-On flag","accounts_in_message":[],"_revision_number":3},{"id":"bf5eea02c90d45805de3962b7f567d6753b3dc23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:31:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f0232efe8d84534a0cdcc06b9ec18fa : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33a0db56d4f14d5caa2c3769d7c8d932 : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37f84b0347984b78bd6bb1085d436c83 : SUCCESS in 4m 50s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e30b2680f59942c89fb16bbf3a6cb680 : SUCCESS in 18m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/194117ad91d64187b85c1502bc37efa0 : SUCCESS in 18m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29b71bc0e3d24273983821e2d677afa3 : SUCCESS in 15m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95fffc6ec89145f8b04c09da5d276255 : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/97f91b1dc758487aaf1b493beb2aee82 : SUCCESS in 53m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f492485a7c5f4769b44a05c9fefe60de : FAILURE in 52m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e75a63c9d33b4fd7b3af59eaca0babc5 : SUCCESS in 47m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/04d33c47603745669eb1d38d7cac9aa9 : SUCCESS in 40m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7aa86e518e454e37ad54a32cd40245fd : SUCCESS in 1h 37m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e64ac9ecf4e5496eae01b3b721254563 : SUCCESS in 1h 58m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e532c8fa78e94d5baa0f7d80fe01c708 : SUCCESS in 1h 46m 16s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b78663e4e42f49a6800a674f7e96e4c9 : SUCCESS in 1h 07m 05s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1e9a11ed456a4ae8919c1406741b9f2a : FAILURE in 14m 26s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/fbf4d4323f584271a17dbf39fa29ac9b : FAILURE in 11m 47s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99e740531d954a4f9b56749b8bc5018a : FAILURE in 8m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f380d5732e504e0ca848ba1ecf6cbb2a : SUCCESS in 5m 11s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a3650d735bf74f7ab0485efeb69f8992 : SUCCESS in 1h 50m 13s","accounts_in_message":[],"_revision_number":3},{"id":"fe086081fff43c8560017163c97dce811158d67f","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-07-23 20:01:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0c4570a93d33454e440cf24b59ff60fe98789a45","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-24 15:15:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"771a74c6004ad549c744df7acf8dd76f8337a3a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 16:58:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40c549143127470885f06ca1a65c969e : SUCCESS in 7m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c60cd02dc79429c9064bfc1406fd10a : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfec3b7fb1e4485aba8e07e832f65022 : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/14c9a1f3097642be9f6c5b7fca7b5d62 : SUCCESS in 16m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/aa6ced3f1011419dae38773a628abafe : SUCCESS in 18m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c297f907efb04b1c8fc3d572ed395813 : SUCCESS in 16m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f2a1f887e264f6db3b6a9298dd2f809 : FAILURE in 8m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8e8776b3ef2e4d1c8a3043cba93fb2c6 : SUCCESS in 56m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3dd39cc483d4231824c888306e182ef : FAILURE in 53m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5e1f8e2ed3043009de95530c6fd588f : SUCCESS in 46m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/527333cc3f6a4e5f86d5a86e4f328987 : SUCCESS in 31m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a650cf7f528e4454b6507234da0b518e : SUCCESS in 52m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca6ef30395c24e64acb975657b195c70 : SUCCESS in 1h 29m 21s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cc53a860c59b487296e04f5a0ab5793b : SUCCESS in 1h 25m 24s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3e9db29ff0c4408995f5f771f4e48831 : SUCCESS in 1h 00m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5edb1bd2a7144ff7bc46de08ffef1b84 : SUCCESS in 22m 21s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/5f76a6a451f145b0a21e390ec346f921 : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5be3ba9474a547bf969bee9a6c53a652 : FAILURE in 7m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74473ee9f3c9402ab9959b6fa049d7fb : SUCCESS in 4m 17s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/60fcd125c835451b895f37f2fdd6c065 : FAILURE in 1h 34m 07s","accounts_in_message":[],"_revision_number":3},{"id":"2c84328c8b32c046c7e1f6fe33cb96a51e826ab3","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-24 18:06:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0ca14fd867617ad19c8c90307245058b20d08a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 19:58:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f14fe6e9e8874b75a707221d241004e5 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fca66ccdbb7944a1be90d4c15158ddcc : SUCCESS in 5m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ddfdc1d540648509945308e29ef4187 : SUCCESS in 5m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bc7b9d14b2e041f19fa9824d7772e438 : SUCCESS in 18m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/23e66f55ed024a7ab8810d8bc52547d1 : SUCCESS in 16m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05187fd7686c4172b5d6b9f103547923 : SUCCESS in 15m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e079c26b2204fa9b8f2dc4939de9372 : FAILURE in 8m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7994133046d440a28de1e9b0edab76ca : SUCCESS in 51m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/681418c9f7a24bfd86709e4b918bf5b3 : FAILURE in 1h 01m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e59514e61e34e4786c772d97913500a : SUCCESS in 49m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9fb59ae76423421bb3717a696a9d58a4 : SUCCESS in 38m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/84b24af6742c452c97cf612a6c05e546 : SUCCESS in 1h 24m 39s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1aced7d5d87d4c32b3c75a5ff95a2adf : SUCCESS in 1h 43m 07s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d168c61547fe475198b3d75c6ed0d3e0 : SUCCESS in 1h 10m 01s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb0900571d2e49cc9468adbd9232858f : SUCCESS in 51m 01s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f751ff9f62324bc09ff2734b92342672 : SUCCESS in 20m 16s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/09b1e49ddd5f479bbffb749691363d1a : FAILURE in 9m 06s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f81ca3ea5b94ca98d081620886e1498 : FAILURE in 7m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a15e2f85759e4a1d8c540346dffd31f1 : SUCCESS in 4m 36s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5595363bb4c14b33b7cd01f88d697fa9 : SUCCESS in 1h 40m 26s","accounts_in_message":[],"_revision_number":3},{"id":"1471b935cc6ae71a92ce59f41eea3151d57acae9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-27 12:49:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7078df85c90e2a7628a657880a74ee03bbc937fe","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-27 17:41:26.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"17544829b71e3efdd8e5b347b00dca9733b831f5","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-27 17:41:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8521a0e43e4c2c135c28152777333ea89d07b3b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:17:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ec62ca236b34bb0aedca9df6e2f9c54 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86ba422ecea64aafba84468917268749 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cdc421a8d62439fabf9b76d17d4d3d3 : SUCCESS in 4m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/36c5167dbb3d47bbbba76bd3f19b513d : SUCCESS in 17m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/df90bbdf2e644801a31d72218e65e0df : SUCCESS in 18m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d36163467b5e46509dfc78e9d78ff97c : SUCCESS in 16m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2b2c98114464c83b239d28c8ef3735a : FAILURE in 10m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f43ad062f0f0489481fa6fe27c72ee08 : SUCCESS in 52m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aef336d04a814e1c9bbdcec802b231cd : FAILURE in 56m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e4f2d35ff524b849153222435f0767c : SUCCESS in 49m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7bf1a22b15fe42a59f019439817b42db : SUCCESS in 39m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/74da28c4aa9e4284a10fd7b769372d75 : SUCCESS in 1h 04m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4186698261784510856b70b8d0eb219b : SUCCESS in 1h 29m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bb75795786a4415392dc4d4ecd115130 : SUCCESS in 1h 41m 13s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aaf3fbf0c345479692bab8c58f36f7bd : SUCCESS in 58m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e95320412b4646119555a1b04ee0d22f : SUCCESS in 20m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/44b5e0ec4bef40f599c6e2fcfc44e836 : FAILURE in 8m 42s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b72852ac499431bafde97c7d8fb5fb5 : FAILURE in 7m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66494f2efa044d9d94669d16b52b87d9 : SUCCESS in 4m 28s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4c7ea4e7fbb74157be70d3314bb94572 : SUCCESS in 2h 20m 55s","accounts_in_message":[],"_revision_number":4},{"id":"fd644b903f9f8afa721b84282d220cc003410912","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-27 20:36:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"61c91a9497b9ab09ca9b93ad0fb546339387d88a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-07-27 20:38:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e308804621ed20fed23cc94a8c1c331adca6f42a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-27 22:34:16.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"5f8920e2d5c7ddb813ee165fdd5a20ff89040700","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-27 22:34:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"350b116956d88db94b1ad9b2ccc6b0a6d2d5bcb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-27 23:22:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"261310d458ed517ae9cfd10eb7cc9dfa6a8e7c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 00:41:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca26d8b193e64ad28c5b18e06bdaeeec : SUCCESS in 4m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c843c3a5fb54fbb9a48328b71b82172 : SUCCESS in 6m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/119a9bdf02df41e2bd25d51bf64cb158 : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d88aa4d0e9144dda92dd36e49d1af6c6 : SUCCESS in 17m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/16bf0dad30e34266998dd7c1a4c98259 : SUCCESS in 13m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98667b66deda441e8566d85b5b21ace0 : SUCCESS in 16m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a84bb0acace6454dbbbf38a28df18c18 : FAILURE in 7m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0baee45c5dc746a7a66ec7f75e4424d0 : SUCCESS in 1h 04m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aad47b42286f4813a3f5f6499a5cf741 : FAILURE in 51m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d38f48b78ffb40058eb1a41b9eef31c2 : SUCCESS in 45m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f7f529f4f1804dd592e17a70dad9a5ed : SUCCESS in 43m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/63246b4f433e4db3af84f84d41076bdc : FAILURE in 1h 06m 41s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fe8f1ad15e1c4e79a1921bace6c5cf24 : SUCCESS in 1h 31m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1a70b319d6d04f04b79a7f341a1ac2ff : SUCCESS in 1h 30m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b4c399f806a942889264817f470cb337 : SUCCESS in 1h 02m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4fcd3902a27844c89ee29fe3686c3347 : SUCCESS in 21m 06s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9788ef3b76a244e19136ab42b0c22808 : FAILURE in 10m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3f61029507f45ff8a77f036ca352cf4 : FAILURE in 7m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e5e91623814910bb764368a7608b07 : SUCCESS in 4m 07s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e011d8495a824c338838339f58cc07d7 : SUCCESS in 1h 54m 54s","accounts_in_message":[],"_revision_number":5},{"id":"bb3b78a75e78138779fcaa9a265339f650a517dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-28 13:56:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"aa582232df821609fe9c43d1f371286aa876ef81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 15:46:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7064c962002e4167b1a0b50dcd42ed4a : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e551861d22ea41728e64263250241697 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58bf67030429499e9473c2d12cc4535b : SUCCESS in 4m 51s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9966453d16d44a08aa2ea798af4ae4a9 : SUCCESS in 17m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1b00fd35d3604ef68894b0eb47ae64c3 : SUCCESS in 20m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ab39e5884944449b34f1a2b29748664 : SUCCESS in 16m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8375f34fd61d4e1480ccd03f2190d8dc : FAILURE in 7m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/739f956b97444569a9157199594aef73 : SUCCESS in 49m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfab32c0dd944988a448555331c53196 : FAILURE in 53m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90eb18acf31c483ca76f6a8c415ca8c9 : SUCCESS in 50m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d679723de2614523a34bdd3b9b6a31fb : SUCCESS in 44m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d396a9b532484a9caf4e75bc8bdf6814 : SUCCESS in 1h 26m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/141276f5d8354336a29e098e3c4cc92d : SUCCESS in 1h 30m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e40d58eb15794243b0310cf40f0ff35d : SUCCESS in 1h 35m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/327660ac6fe54b66b38c51f68e312fa0 : SUCCESS in 56m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a3b23b9232b547259cfec7ae225b1bd1 : SUCCESS in 19m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9c39f3660cf34458a84ef0eaf280a6cc : FAILURE in 14m 40s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b305adf131e4cdeb19332f9ca3cc2b1 : FAILURE in 10m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3851e0664bfd4d27be01f83e4a50c678 : SUCCESS in 7m 39s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cb4766ed373a432ebaef2b984f14e64d : SUCCESS in 1h 42m 59s","accounts_in_message":[],"_revision_number":5},{"id":"56659701e3c1950d74b1dd15036803b464b6eed3","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-07-28 15:53:54.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2c435ddbcca142757340c8cc46fd1774508073ff","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-29 19:40:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7e0d892c5be680cd9ceecbdde528c42978154be7","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-07-29 19:43:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d49c38e23411bfaffa001354e9fbb118178be1e9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-30 22:35:14.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b744215d45d06067c274a31aa389855fc5739446","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-30 22:37:55.000000000","message":"Patch Set 5: -Workflow\n\nHolding on +A so that we can have a final look after cinder patch is merged","accounts_in_message":[],"_revision_number":5},{"id":"0f6cefb8bccfc034c2ed216d89feb3109b50fb46","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-08-02 23:31:48.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"db734e0e50fa3a994a428f7dfe7e13e3e8ea4617","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-08-05 22:39:43.000000000","message":"Patch Set 5:\n\nThe review it is depending on is merged. This is ready to go","accounts_in_message":[],"_revision_number":5},{"id":"dae8ddf26977ccf8c00489e4cc6fd9e0f1bf8938","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-06 00:16:58.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"47d9d5884945f443ca01eea2a8d5e5d694f6069c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 00:17:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6714a7ef643f3a0fcad44e3f26a8860da8bde3f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 02:04:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8319b4afeb5844b98fc72cd896b12d77 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/620017c7cdd247068e4ca2a0a5c4cef9 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3287ba332ada4561b4585e17e09e5c6a : SUCCESS in 3m 57s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/990c1cbcb19c4048ac2fc86ab6dc53b1 : SUCCESS in 13m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/27a503545062458991a21a7894cb776a : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e2faa8f52a94a92a85897d9cce389be : SUCCESS in 12m 23s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b16c579c88a94536934890f4386c2854 : SUCCESS in 1h 11m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/61dee115e50d4e7b8ca6a4e9fa09a73c : SUCCESS in 1h 25m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b2a545eb7bb4e7f89c51455cd15f545 : SUCCESS in 1h 16m 12s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e6fe769903c485fba6fe9e4ffd1d8d3 : SUCCESS in 1h 05m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0d4378d7540c43ca8d2ef0b2fc010d16 : SUCCESS in 17m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/252508fcafee43f5a98d40156894aa42 : SUCCESS in 3m 09s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1cce01fdf15b4af1b729f7182e7d9669 : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":5},{"id":"87de96ce3e0df3579987e09d1a4be2f1d8d58d2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 02:04:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"fe1707b1dd25f8d76bc451e5ac7b236a5e9c5edf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 02:05:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b2dbad26591455c9741a756f9177d1e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bd7576e518dbee753dfa99722a0e749c72380308","revisions":{"8d7fe22bd8854be3e3836169ab733a45e45d76e7":{"kind":"REWORK","_number":1,"created":"2020-07-20 19:39:57.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/16/742016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/16/742016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/1"}}},"commit":{"parents":[{"commit":"d8a421fc48192966fd6fcc06d4be010424f3d8b5","subject":"Merge \"Change to relative include\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d8a421fc48192966fd6fcc06d4be010424f3d8b5"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-20 19:38:16.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-20 19:38:16.000000000","tz":-300},"subject":"Rebranding of VxFlex OS driver to PowerFlex","message":"Rebranding of VxFlex OS driver to PowerFlex\n\nVxFlex OS driver is rebranded to PowerFlex.\nWill deprecate the VxFlexOS template in\na new patch\n\nChange-Id: I831cd341a838280bdd91bdca94af1a860cdc8006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8d7fe22bd8854be3e3836169ab733a45e45d76e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8d7fe22bd8854be3e3836169ab733a45e45d76e7"}]},"branch":"refs/heads/master"},"4de7589b262ba0c541bcab04443155d8c5179cb2":{"kind":"REWORK","_number":2,"created":"2020-07-20 20:00:28.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/16/742016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/16/742016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/2"}}},"commit":{"parents":[{"commit":"d8a421fc48192966fd6fcc06d4be010424f3d8b5","subject":"Merge \"Change to relative include\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d8a421fc48192966fd6fcc06d4be010424f3d8b5"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-20 19:38:16.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-20 20:00:21.000000000","tz":-300},"subject":"Rebranding of VxFlex OS driver to PowerFlex","message":"Rebranding of VxFlex OS driver to PowerFlex\n\nVxFlex OS driver is rebranded to PowerFlex.\nWill deprecate the VxFlexOS template in\na new patch\n\nChange-Id: I831cd341a838280bdd91bdca94af1a860cdc8006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/4de7589b262ba0c541bcab04443155d8c5179cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/4de7589b262ba0c541bcab04443155d8c5179cb2"}]},"branch":"refs/heads/master"},"640d9972cf82a55745b4bdf099db4d76d91f348e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-22 16:51:48.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/16/742016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/16/742016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/3"}}},"commit":{"parents":[{"commit":"d8a421fc48192966fd6fcc06d4be010424f3d8b5","subject":"Merge \"Change to relative include\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d8a421fc48192966fd6fcc06d4be010424f3d8b5"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-20 19:38:16.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-22 16:50:08.000000000","tz":-300},"subject":"Rebranding of VxFlex OS driver to PowerFlex","message":"Rebranding of VxFlex OS driver to PowerFlex\n\nVxFlex OS driver is rebranded to PowerFlex.\nWill deprecate the VxFlexOS template in\na new patch\n\nDepends-On: https://review.opendev.org/#/c/740522/\nChange-Id: I831cd341a838280bdd91bdca94af1a860cdc8006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/640d9972cf82a55745b4bdf099db4d76d91f348e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/640d9972cf82a55745b4bdf099db4d76d91f348e"}]},"branch":"refs/heads/master"},"1a1d058a3c85f688af813397efa1213501683701":{"kind":"REWORK","_number":4,"created":"2020-07-27 17:41:26.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/16/742016/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/16/742016/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/4"}}},"commit":{"parents":[{"commit":"d8a421fc48192966fd6fcc06d4be010424f3d8b5","subject":"Merge \"Change to relative include\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d8a421fc48192966fd6fcc06d4be010424f3d8b5"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-20 19:38:16.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-27 17:40:32.000000000","tz":-300},"subject":"Support for PowerFlex cinder backend","message":"Support for PowerFlex cinder backend\n\nVxFlex OS driver is rebranded to PowerFlex.\nThis patch adds support for PowerFlex\nWill deprecate the VxFlexOS template in\na new patch\n\nDepends-On: https://review.opendev.org/#/c/740522/\nChange-Id: I831cd341a838280bdd91bdca94af1a860cdc8006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/1a1d058a3c85f688af813397efa1213501683701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/1a1d058a3c85f688af813397efa1213501683701"}]},"branch":"refs/heads/master"},"bd7576e518dbee753dfa99722a0e749c72380308":{"kind":"REWORK","_number":5,"created":"2020-07-27 22:34:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/742016/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/16/742016/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/16/742016/5"}}},"commit":{"parents":[{"commit":"d8a421fc48192966fd6fcc06d4be010424f3d8b5","subject":"Merge \"Change to relative include\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d8a421fc48192966fd6fcc06d4be010424f3d8b5"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-07-20 19:38:16.000000000","tz":-300},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-27 22:33:58.000000000","tz":0},"subject":"Support for PowerFlex cinder backend","message":"Support for PowerFlex cinder backend\n\nVxFlex OS driver is rebranded to PowerFlex.\nThis patch adds support for PowerFlex\nWill deprecate the VxFlexOS template in\na new patch\n\nDepends-On: https://review.opendev.org/#/c/740522/\nChange-Id: I831cd341a838280bdd91bdca94af1a860cdc8006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/bd7576e518dbee753dfa99722a0e749c72380308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/bd7576e518dbee753dfa99722a0e749c72380308"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
