)]}'
{"id":"openstack%2Fpuppet-cinder~785242","triplet_id":"openstack%2Fpuppet-cinder~master~I6f4c6a51f8e33616069411a926a63ba2b388e25d","project":"openstack/puppet-cinder","branch":"master","topic":"debian-uwsgi-support","hashtags":[],"change_id":"I6f4c6a51f8e33616069411a926a63ba2b388e25d","subject":"Add support for cinder_api_uwsgi_config in Debian","status":"MERGED","created":"2021-04-07 17:35:49.000000000","updated":"2021-04-14 16:22:12.000000000","submitted":"2021-04-14 16:19:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785242-debian-uwsgi-support","meta_rev_id":"96b36fec01d783c152aa40146e82b5b70510fcf7","_number":785242,"virtual_id_number":785242,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 16:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"all":[{"value":1,"date":"2021-04-12 06:50:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":2,"date":"2021-04-14 14:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-12 16:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-12 06:55:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":1,"date":"2021-04-12 06:46:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":1,"date":"2021-04-14 14:07:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 19:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 06:46:02.000000000","updated_by":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"reviewer":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"state":"REVIEWER"},{"updated":"2021-04-12 06:50:18.000000000","updated_by":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"reviewer":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"state":"REVIEWER"},{"updated":"2021-04-12 06:55:05.000000000","updated_by":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"reviewer":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"068677a04b05edcd9d3e41bba83db7e2a6c5d936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 17:35:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8086431ab5cead20ea2cb7a5c212a5d4c0ef6399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 19:23:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/122db85e35244864b27b2561b7bd58c7 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56881120f4bc409ca2c48dbc8c804973 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e2e6cf736ee41bda32412c050d484fa : SUCCESS in 11m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa00ee923d1d4985a3ed914bdbd92ccf : SUCCESS in 10m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00a8454e27c24c12b857b80d8e9830e6 : SUCCESS in 9m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2252b57583c64cb284a151c7cb6640a6 : SUCCESS in 49m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67769b049238468387e6aaf20d4e4d1c : SUCCESS in 54m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b42c5d884734ce4ba8a291008e53901 : FAILURE in 30m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb2076e511b74d5c8ff8cdb75fbdd3d7 : SUCCESS in 36m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6f33030b52a497ba3a6565c847004bf : SUCCESS in 1h 40m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ded0a5751ed4a3586b5b7e845f30556 : SUCCESS in 1h 27m 37s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f152f7cb4b94d17a7eb0aa75bddd641 : SUCCESS in 1h 28m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eef336b17dea49d892ab794b0ee640e6 : SUCCESS in 57m 20s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fa2250e75be49d784a68c4cf2d83152 : SUCCESS in 18m 48s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fcbded52ca540bfa661a324d49233b1 : SUCCESS in 14m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec744a35517c43c59c5ce97068a6a163 : SUCCESS in 4m 19s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/95d72c2842dc425596880e8fb75c8c94 : FAILURE in 1h 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"eb24268e11862ae7210b759e3f278c421a13f872","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 21:55:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bec1105904f7790fd9ba9c14768683cbcd1dfc2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:38:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff7686756c984a1083d465ff133bf1cf : SUCCESS in 12m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f888cfc6a46949ff8ff04dad2cedf61e : SUCCESS in 4m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c7391de91164119bcbb956aa7d7e76e : SUCCESS in 12m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8946defb4dc241d4bed1d7c35eb27c50 : SUCCESS in 11m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bb8ec0ec4414051a32352ed2c80495b : SUCCESS in 10m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74f36afac3364c5483ba85fd24f3e90c : FAILURE in 37m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d69cb6a366004f5e97f11351e34c8ec2 : SUCCESS in 53m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/339c78d6bf07401789c80b8b806bcda9 : FAILURE in 40m 18s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9465a92693a547dcbcdd74fca7dbf2ed : SUCCESS in 43m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e47bac586cf4a25ab561ecd0fcf3969 : SUCCESS in 1h 19m 51s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ac6bade49c64bfbbef8b5ef6473eb9a : SUCCESS in 1h 18m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0826d62bcd34bc087eeecdd8aafb5b7 : SUCCESS in 1h 14m 27s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35cf6806f0604df5ae73da4446fb2c4e : SUCCESS in 59m 58s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f1a126e26f1415ab6c47d0fae937feb : SUCCESS in 18m 53s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c56a5a06c614183865b9664745e7d84 : SUCCESS in 16m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c7192199c324820b1684b91783d49e7 : SUCCESS in 4m 43s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1834185a622e4305896c20a0b71c1689 : FAILURE in 1h 35m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0418d0d97c434ad36980296388d6fad3dba6865b","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-08 07:52:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ad17fcfe0b92548598c5ac9c3fbee2409b4d80d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:16:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5fa2b7cb18140a2a02bf5b469d0ba94 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b193f7900d1457f9ef63ac7564104f2 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae38df3a0eb8488faa0843e2546258e0 : FAILURE in 8m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c50b3e1640414d3785f01e2fbcd1b2f6 : FAILURE in 5m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90112e32d56c487b834c256cc2841276 : FAILURE in 5m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8f183515a514918bd252482901be47d : FAILURE in 17m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1425c12a0ce74f448bc24688f40f3a2a : FAILURE in 8m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e8d941bea114ded98d400172c43828b : FAILURE in 28m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7b7f89033564185b3eae296649c3948 : SUCCESS in 37m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/26d73ebfa6e74420bb48919c8d511419 : FAILURE in 13m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a83a5d75eaf42d392b06a043454739d : FAILURE in 9m 41s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/21871e6333c840cb889679054032e467 : FAILURE in 1h 14m 34s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3570640198ee4109a685120523a3fd65 : SUCCESS in 59m 28s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7e82db59714442caf637370170de185 : FAILURE in 11m 32s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4104075a5245486d83da0261a464f101 : FAILURE in 6m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d066f41f838a476c993380050a2bc5a9 : SUCCESS in 5m 11s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8b8aa5127cdd44a2825fc3bd24d57734 : FAILURE in 1h 18m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c136a54f8c90c2a5a0859f27ed9315137aba6af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-09 21:46:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4fd6da96fe12018d1b4b915469be605c5e2d586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-09 22:57:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98ae27580879f47b1b572f7a29d91687e1d38f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 01:26:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8375cd9e22040099be5a2fb2eb25d5c : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/523850847fd64a76919d2f697943a204 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/536a7d5fbb95499e8f8cbb1d28338f9b : FAILURE in 7m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96b8496abd054edb9f30802622efad61 : FAILURE in 5m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7301c8eadf354657955a9c8b92c628fe : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceb603c94ef943b1aca62759b4dd798f : FAILURE in 34m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba74317eb5744e5cad29751836fb6f6b : SUCCESS in 57m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9835cfe930a4fef8cf6c53d575d9995 : FAILURE in 31m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96e3031edd824db99e8965064912ca53 : SUCCESS in 51m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2e34170e9ef4a8c9ef8001801a95ec8 : SUCCESS in 1h 06m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f273f8b3dda24d74ad20211354b08e6f : SUCCESS in 1h 18m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b30f038027de442db600fd426bb652e8 : SUCCESS in 1h 15m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb19ed9668df4f1cb7bc0217d2086265 : SUCCESS in 1h 02m 55s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c364c973c6c345c4b22f741b3ef98f6b : SUCCESS in 16m 49s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/599217428a7e47c49ef231655a699e6c : SUCCESS in 14m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d56c56f881cd4e64bd1b8c35087a0e8b : SUCCESS in 4m 06s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/67bb3983a59e47fea529f39d3a74a4c6 : SUCCESS in 2h 23m 29s","accounts_in_message":[],"_revision_number":3},{"id":"4ff89846ff52e5fcfc52f9350a1c0e4549e51e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-10 08:32:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"315d4c5d76d8bb3e6bdab07845079e1bc3babb41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 11:31:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/236bfa52cf1f4dadb1a4f1928c80a78e : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eba79521a8f5441ea3ecc03263502479 : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/997f9c68708c41d184801f087629609a : FAILURE in 7m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7573f394880640ed9d1660f9debd37c8 : FAILURE in 6m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f19fc5abc0d426cacf33abebe70e82d : FAILURE in 6m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a215a9f8b3334242b9a7e85cca894974 : SUCCESS in 1h 05m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ff461a4af19419699197383016f7d14 : SUCCESS in 58m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e09e5bc29b624b9e9f47c2baca951525 : FAILURE in 37m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05b94fee090f4ae5884ff93844de534a : SUCCESS in 29m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ccbb247ced949f4a6db8a26056d8684 : SUCCESS in 52m 26s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d190f63bc4a433e8c96eba86188c553 : SUCCESS in 57m 13s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3d67e01f26e4814a8da0dc87157f802 : SUCCESS in 57m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/499e4134a80a4121b362296eeb1cdff0 : SUCCESS in 41m 06s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3750b5158cba40bcaa1d34faa7481704 : SUCCESS in 13m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab6f78c90f694ea795b0d7565651ac6c : SUCCESS in 11m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61f732c018974dd98bd5f0f5f144aaae : SUCCESS in 3m 49s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e2bcffbc59bb4fa59aedb08c3c74584b : SUCCESS in 2h 52m 25s","accounts_in_message":[],"_revision_number":4},{"id":"67da11998ddeb5c767fc8384a7b799763df2dc6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-10 16:21:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8d497452f2583170b995feabbb03f517f4d8c32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 18:26:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d377252d3c3443ef8b042b4bb9eb8ef4 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0253c2a138a14903bc10d3a2493301a9 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/02ac66c199bc4a64b41afc4aa29ca216 : FAILURE in 7m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df88a167e4fd450cb7740b19b74f9f80 : FAILURE in 5m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d69a527156848fe81becce8284fc068 : FAILURE in 5m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a19ec32a7ecd4e428ed4ada2adb030ff : SUCCESS in 55m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/828bce4bf86b4ad7935c88a39255789d : SUCCESS in 55m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd995b4d9ac34f86af544a3acc6717c6 : FAILURE in 28m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d22a614f122d4c9f8bc56219fc45143a : SUCCESS in 37m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a20a2b4394f40af95801b19812a6dbd : SUCCESS in 1h 22m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0de64756b2c044c6a6f46792f086d757 : SUCCESS in 1h 30m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/45795cabdd454d56bfc344ebef817feb : SUCCESS in 1h 27m 43s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ceeb719545cd41568b572e55db52c04f : SUCCESS in 58m 49s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c6b91992d674f6e97c7550fe5de9d53 : SUCCESS in 19m 12s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab13b492761d4b62a10c93ad50096b06 : SUCCESS in 14m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d40a6dbbd5ce4a219550ba086ef19280 : SUCCESS in 3m 55s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f8275d08f8d14ae681bfaa85a456dd7a : SUCCESS in 2h 03m 23s","accounts_in_message":[],"_revision_number":5},{"id":"3bbc53961dc3451cfca1da3815b5405d7fc6f399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-10 20:41:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"93d3ec64659ee86bf5babfc9a15b575c0d50040a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 22:39:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/000f6aa851f04886b2a59ea546fec88c : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/649283e58fd54038ba7ee563836148d4 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8191700a176546f7843f53f29b97537a : SUCCESS in 12m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ca6a08ab9294bb1b4334b6c84d4192a : SUCCESS in 8m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a243bbecb9a54cdf8e8954f113c41317 : SUCCESS in 8m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cb071f5a21c4dbca7205b6ae8e71ea6 : SUCCESS in 48m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/811efaf5789341cfb18355fb10cc22da : SUCCESS in 47m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b7e049d64ae44b0815257f501321111 : FAILURE in 25m 54s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfc1836f521248c4ad40e710d8394b8c : SUCCESS in 40m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d67b312b9fd4f2e9246fed45970baea : SUCCESS in 1h 04m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ff84d15966548d4b95b157a2f58c55e : SUCCESS in 54m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b651757b53b24dcf842b21397287b39a : SUCCESS in 58m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4b9201fd12b430b80b2e616da8c3bf3 : SUCCESS in 39m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/9de47da7bb6f4ec6bb657905493a26b8 : SUCCESS in 14m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b8d1c41ccc14ad79771436502bae8c4 : SUCCESS in 15m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29f622acdfa54fa683eb6d00c3f90531 : SUCCESS in 4m 11s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d5d8bbe88e234682938620a2f3b56bb0 : SUCCESS in 1h 57m 11s","accounts_in_message":[],"_revision_number":6},{"id":"c29eb234cfaee5482d41d738014771da2a17be63","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2021-04-12 06:46:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2d6c78ebe9b67b9269d1ac4e51e1ec0fe0c305e1","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2021-04-12 06:50:18.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"25cdaeda6c520a904562b385ef0d96d7f6565c0d","author":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"date":"2021-04-12 06:55:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"ab79b283bcb09c0edde58a574921c7c98420cb08","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-12 16:49:15.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"187636af1a9140800ea8b144d942ee5edc68aa11","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-14 14:07:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c74a501f70ecda84632efbbec027730f55953cb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:08:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"047d7519e2ab109a1bce063c3ec3cffbcec712c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:19:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55cc9e8a4bc94c14955db675f321fa0f : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc3dcb264621480e90914f1ea5479de5 : SUCCESS in 5m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3337fe70ebac4164a3f4043d2619b950 : SUCCESS in 15m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2619983157ea40028cb6f295dba98d0f : SUCCESS in 12m 35s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5e6516d10d24dafb7a36b841aa324ca : SUCCESS in 1h 40m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8810be96cf8f4a34b02ab34b52819f6f : SUCCESS in 1h 32m 45s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f19ce92bd1c542bea9e94ae9cca6bf81 : SUCCESS in 1h 09m 05s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/97841548ad4e4acba4d60d44b02e3edb : SUCCESS in 1h 04m 43s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8cdd8e4c2324ffbae0828e4970937dd : SUCCESS in 18m 50s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e154b8d381d14f9d84e96fbe7acdda2c : SUCCESS in 15m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7b835c9178a43e2a38c7518ebff8005 : SUCCESS in 4m 01s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3a5ea925222741d8aca7114359c08337 : SUCCESS in 2h 05m 52s","accounts_in_message":[],"_revision_number":6},{"id":"41747b152ac1f07c4d515ef799bfaa297fc2d16c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:19:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"96b36fec01d783c152aa40146e82b5b70510fcf7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:22:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ea0778ffe3c4db69ecbbe01836196fe : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"003b8d6ab69201c7ebe179f48b8764bea34b53e4","revisions":{"df04e91ebb01ea50606497865e9a97ac811c74f9":{"kind":"REWORK","_number":1,"created":"2021-04-07 17:35:49.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/42/785242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/42/785242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/1"}}},"commit":{"parents":[{"commit":"930d112c8e22291d2716a4585a8502d7eb8eb103","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/930d112c8e22291d2716a4585a8502d7eb8eb103"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:34:36.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:34:36.000000000","tz":120},"subject":"Add support for cinder_api_uwsgi_config in Debian","message":"Add support for cinder_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Cinder API. Therefore, this patch adds a new\ncinder_api_uwsgi_config provider as well as a new\ncinder::wsgi::uwsgi class.\n\nChange-Id: I6f4c6a51f8e33616069411a926a63ba2b388e25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/df04e91ebb01ea50606497865e9a97ac811c74f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/df04e91ebb01ea50606497865e9a97ac811c74f9"}]},"branch":"refs/heads/master"},"bec6bff462c604afb1f4c91f7dfc2ef9bc5c56c0":{"kind":"REWORK","_number":2,"created":"2021-04-09 21:46:38.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/42/785242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/42/785242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/2"}}},"commit":{"parents":[{"commit":"930d112c8e22291d2716a4585a8502d7eb8eb103","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/930d112c8e22291d2716a4585a8502d7eb8eb103"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:34:36.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-09 21:46:27.000000000","tz":120},"subject":"Add support for cinder_api_uwsgi_config in Debian","message":"Add support for cinder_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Cinder API. Therefore, this patch adds a new\ncinder_api_uwsgi_config provider as well as a new\ncinder::wsgi::uwsgi class.\n\nChange-Id: I6f4c6a51f8e33616069411a926a63ba2b388e25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/bec6bff462c604afb1f4c91f7dfc2ef9bc5c56c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/bec6bff462c604afb1f4c91f7dfc2ef9bc5c56c0"}]},"branch":"refs/heads/master"},"78789bc02ddbbf33fa43d0523714e0f86806b33d":{"kind":"REWORK","_number":3,"created":"2021-04-09 22:57:22.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/42/785242/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/42/785242/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/3"}}},"commit":{"parents":[{"commit":"930d112c8e22291d2716a4585a8502d7eb8eb103","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/930d112c8e22291d2716a4585a8502d7eb8eb103"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:34:36.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-09 22:57:09.000000000","tz":120},"subject":"Add support for cinder_api_uwsgi_config in Debian","message":"Add support for cinder_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Cinder API. Therefore, this patch adds a new\ncinder_api_uwsgi_config provider as well as a new\ncinder::wsgi::uwsgi class.\n\nChange-Id: I6f4c6a51f8e33616069411a926a63ba2b388e25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/78789bc02ddbbf33fa43d0523714e0f86806b33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/78789bc02ddbbf33fa43d0523714e0f86806b33d"}]},"branch":"refs/heads/master"},"3b9adc53e9352bd9880b9d6989fa69d103691c3f":{"kind":"REWORK","_number":4,"created":"2021-04-10 08:32:45.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/42/785242/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/42/785242/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/4"}}},"commit":{"parents":[{"commit":"930d112c8e22291d2716a4585a8502d7eb8eb103","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/930d112c8e22291d2716a4585a8502d7eb8eb103"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:34:36.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-10 08:27:55.000000000","tz":120},"subject":"Add support for cinder_api_uwsgi_config in Debian","message":"Add support for cinder_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Cinder API. Therefore, this patch adds a new\ncinder_api_uwsgi_config provider as well as a new\ncinder::wsgi::uwsgi class.\n\nChange-Id: I6f4c6a51f8e33616069411a926a63ba2b388e25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3b9adc53e9352bd9880b9d6989fa69d103691c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3b9adc53e9352bd9880b9d6989fa69d103691c3f"}]},"branch":"refs/heads/master"},"40cfc2d3ce675f001cd078f1cea3caf930d00ad1":{"kind":"REWORK","_number":5,"created":"2021-04-10 16:21:40.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/42/785242/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/42/785242/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/5"}}},"commit":{"parents":[{"commit":"930d112c8e22291d2716a4585a8502d7eb8eb103","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/930d112c8e22291d2716a4585a8502d7eb8eb103"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:34:36.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-10 16:21:26.000000000","tz":120},"subject":"Add support for cinder_api_uwsgi_config in Debian","message":"Add support for cinder_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Cinder API. Therefore, this patch adds a new\ncinder_api_uwsgi_config provider as well as a new\ncinder::wsgi::uwsgi class.\n\nChange-Id: I6f4c6a51f8e33616069411a926a63ba2b388e25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/40cfc2d3ce675f001cd078f1cea3caf930d00ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/40cfc2d3ce675f001cd078f1cea3caf930d00ad1"}]},"branch":"refs/heads/master"},"003b8d6ab69201c7ebe179f48b8764bea34b53e4":{"kind":"REWORK","_number":6,"created":"2021-04-10 20:41:08.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/42/785242/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/42/785242/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/42/785242/6"}}},"commit":{"parents":[{"commit":"930d112c8e22291d2716a4585a8502d7eb8eb103","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/930d112c8e22291d2716a4585a8502d7eb8eb103"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:34:36.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-10 20:40:57.000000000","tz":120},"subject":"Add support for cinder_api_uwsgi_config in Debian","message":"Add support for cinder_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Cinder API. Therefore, this patch adds a new\ncinder_api_uwsgi_config provider as well as a new\ncinder::wsgi::uwsgi class.\n\nChange-Id: I6f4c6a51f8e33616069411a926a63ba2b388e25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/003b8d6ab69201c7ebe179f48b8764bea34b53e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/003b8d6ab69201c7ebe179f48b8764bea34b53e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
