)]}'
{"id":"openstack%2Fpuppet-cinder~788315","triplet_id":"openstack%2Fpuppet-cinder~master~I9af225f714bf2df6b4c579ff12392d5c30769cb1","project":"openstack/puppet-cinder","branch":"master","topic":"vrts","hashtags":[],"change_id":"I9af225f714bf2df6b4c579ff12392d5c30769cb1","subject":"Remove support for the Veritas HyperScale Driver","status":"MERGED","created":"2021-04-27 15:22:30.000000000","updated":"2021-07-07 22:04:04.000000000","submitted":"2021-07-07 22:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788315-vrts","meta_rev_id":"badedc4992e5878581a0440395669389abb1173a","_number":788315,"virtual_id_number":788315,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 22:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2021-07-07 13:30:57.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-07-07 20:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-07-07 20:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 15:32:15.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-04-27 18:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 20:01:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ee704eb79fe6b724d23a88675ec256b844e1f426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 15:22:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0de4f9d1c98dabb102b948ce10c9a3dc6792427d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-04-27 15:32:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"405258ac1183e9096c84dd8c9c12e311fae8de52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 16:09:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd748469d08487053f6b7ea48329610fd1c13e58","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 16:11:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36f57afb1dd04491647c57f3a53d27d1fb474c50","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 16:12:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fe5f20119aed68912e87f566010533aa8e7e9ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:19:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6341f9b7f478484fb88e10970c237921 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1106ac13b32c4af08f386233500571b4 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/018c17b4201d4aa697e5f2df97765bb1 : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e24eddc529da47c686740de439713dca : SUCCESS in 10m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2927b2855a824c26bf1bc050d85d5d38 : SUCCESS in 10m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7125a3e1464492f9a984260f473137a : SUCCESS in 51m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e2fcae1de4b4c10902f497dd31eab25 : FAILURE in 1h 01m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/babcca6e7d5a407382d70e4f20403f65 : FAILURE in 23m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b29f743114df4905a14194f8b215276c : SUCCESS in 38m 29s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3c1ac646e644fff9ebc1adcdde0c84a : SUCCESS in 17m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f106bb9a8a87476eb29622e04b5fb519 : SUCCESS in 1h 04m 49s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2fd88f11fcd48be8de850f5104eb872 : FAILURE in 1h 34m 16s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed7eb0d383aa45bdb1661c8ab42b0ee4 : TIMED_OUT in 2h 02m 09s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/37abc1caf1364ef68f624690522976d3 : SUCCESS in 45m 40s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/576f30c9a50442f695479f860996212e : SUCCESS in 22m 01s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c13a4cd142345acbbfd6104510efca6 : SUCCESS in 35m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/515e1fed957a40d791d4082605fe2dde : SUCCESS in 4m 34s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/25f306af80f24b13a15e8c7b4b86a6f1 : SUCCESS in 1h 33m 05s","accounts_in_message":[],"_revision_number":2},{"id":"31a9fcb4548ab7761471cb1a217aa1c64b83b1bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-07 03:52:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2a1e9a3da104613944c3c61f72faae8c2cbcbcec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 05:49:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb92a764ebde4d33851c994eaeec5b9d : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78dc8ab6d051411ab97c2a41b0cc5261 : SUCCESS in 6m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7005ae812caf442086772519e1c69656 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/25e0134303c74b1b966a976406e9a81d : SUCCESS in 13m 32s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9f1fbd7a69f4804ad30306e6c1ce7a2 : SUCCESS in 14m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5bfdf7b25a14a79bd4997e23f5cf212 : SUCCESS in 11m 33s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d77f13e12ae942f187508145b2f8bdc2 : SUCCESS in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32ed6dc31fbd4b5db010065608f8dd3f : SUCCESS in 10m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea6eaf88ece24c928c524491bfacdb87 : FAILURE in 37m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/726c10fe50fa47ebbc01f302c7058fa0 : FAILURE in 42m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/669bc9afd13948bcb9068cc6c3a34b98 : FAILURE in 29m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a614dd3763d41dba9c8ff7ea19e7c46 : FAILURE in 29m 16s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4326d6fe2a2a42a7b1c035d72fef0712 : SUCCESS in 26m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/424e47d5f0b04d0c90a9b334e3c80708 : SUCCESS in 1h 24m 19s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0216db8258c432089044ee21057bcb6 : SUCCESS in 1h 32m 29s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/45ac13dfd2e54598905f8837381dc381 : SUCCESS in 1h 32m 42s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/18315326ba874adda07341c326991316 : SUCCESS in 1h 01m 40s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c8afb38e0cd40d19e45622c2c8f3e34 : SUCCESS in 23m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/97bd5c59aa2a4740905ed70dbc65aecf : SUCCESS in 1h 19m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d90ab77051cd45e283e7a35432643b76 : SUCCESS in 1h 31m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/85971c77a07b4da7924f98fdf05ee166 : SUCCESS in 1h 18m 01s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9a8a1e01c954d12901b5078374619fe : SUCCESS in 1h 14m 41s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab7aefefdfb248b4abf03d9c1b7d26bd : SUCCESS in 22m 14s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c1fc2b58b894e688c567029d0b49515 : FAILURE in 14m 06s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a54502b556164e04aed0b46dd9b8f17e : SUCCESS in 22m 07s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7153d7d18eeb44f09bc5fd6eedc70b95 : FAILURE in 14m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4457c0a77254e9b8e3759b48cc70af0 : FAILURE in 14m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2b1bfeb5890455992310f69b3ab85f8 : SUCCESS in 4m 02s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6ca4001b44e34f6c8bacd33dbfe4f4bf : SUCCESS in 1h 53m 17s","accounts_in_message":[],"_revision_number":3},{"id":"e44686df830655a77bd2b29fd1c93c6dca1df358","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-07 06:39:50.000000000","message":"Patch Set 3:\n\nNote: Usage in puppet-tripleo has been already removed. https://review.opendev.org/c/openstack/puppet-tripleo/+/787590","accounts_in_message":[],"_revision_number":3},{"id":"c8dbe874138a1996badcd69ee6c5d03faac084c6","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-07-07 13:30:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a3a81296c3369bf0079f03b444403e0a48f7b522","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-07 20:01:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8d0e0002ec4a08800e3d87c8009b9e6c7581f565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:02:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d481808f4eccb28d1bdb1c5f0a76c7fac9d7f3a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 22:02:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8245dfed29404ac0a03e76d75d536245 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a0d2972495147e3acde208d8cd9c262 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c987500e8e14cc3bca3980a9cd7e872 : SUCCESS in 4m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/68057a4b882a48acb52f613bebfcfa6d : SUCCESS in 16m 42s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bc4afa1b43d4a2c9b34504fb1d32503 : SUCCESS in 13m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0cc6911ff8f44368da3a1f149db1d35 : SUCCESS in 11m 05s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d5c2883d4664a35ba1e9375e54c6396 : SUCCESS in 11m 26s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/356c9a57cba4494caea66909349d5f26 : SUCCESS in 1h 17m 23s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e55e40ffc20474db6f435492ce9ad34 : SUCCESS in 1h 26m 37s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/22a077cc869e414688dbdf144c593a5e : SUCCESS in 1h 49m 04s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f8afdaf47eb465babc3068b5c8ccbb4 : SUCCESS in 1h 01m 57s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4ae3a7bc50f49eda14236c8448360b7 : SUCCESS in 23m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12ea4c4eedee4e77a09ae8c25a26fba7 : FAILURE in 17m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6113c13959984c0b8459af82bcbf3ece : SUCCESS in 4m 16s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/babd319a523a43e3bf1dc7b7c1d7d9a6 : SUCCESS in 1h 44m 06s","accounts_in_message":[],"_revision_number":3},{"id":"89f574517559e3d941b80697592c63e1885fb46a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 22:02:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"badedc4992e5878581a0440395669389abb1173a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 22:04:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e00fc7e0e4404bfc94da10a03d0b1d73 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"75949efb1d6af18ea971b9b84bf1ad52879cad41","revisions":{"8ec8c1d16116e62f4b35f912e511c1ac121621ab":{"kind":"REWORK","_number":1,"created":"2021-04-27 15:22:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/788315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/15/788315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/1"}}},"commit":{"parents":[{"commit":"3fc60e2b9e1dbdb066d14735925173fb812bbdaf","subject":"Merge \"Improve cinder_type properties regex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3fc60e2b9e1dbdb066d14735925173fb812bbdaf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 15:21:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 15:22:11.000000000","tz":540},"subject":"Remove support for the Veritas HyperScale Driver","message":"Remove support for the Veritas HyperScale Driver\n\n... because it was deprecated during Victoria cycle[1] and driver\ncode was already removed from cinder.\n\n[1] 0da2b260eb6961222a23f7728c41593bcf349030\n\nChange-Id: I9af225f714bf2df6b4c579ff12392d5c30769cb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8ec8c1d16116e62f4b35f912e511c1ac121621ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8ec8c1d16116e62f4b35f912e511c1ac121621ab"}]},"branch":"refs/heads/master"},"6f949003600a8f4149e43b97dac9ffafbbdf57c4":{"kind":"REWORK","_number":2,"created":"2021-04-27 16:09:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/788315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/15/788315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/2"}}},"commit":{"parents":[{"commit":"3fc60e2b9e1dbdb066d14735925173fb812bbdaf","subject":"Merge \"Improve cinder_type properties regex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3fc60e2b9e1dbdb066d14735925173fb812bbdaf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 15:21:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 16:09:39.000000000","tz":540},"subject":"Remove support for the Veritas HyperScale Driver","message":"Remove support for the Veritas HyperScale Driver\n\n... because it was deprecated during Victoria cycle[1] and driver\ncode was already removed from cinder.\n\n[1] 0da2b260eb6961222a23f7728c41593bcf349030\n\nChange-Id: I9af225f714bf2df6b4c579ff12392d5c30769cb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/6f949003600a8f4149e43b97dac9ffafbbdf57c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/6f949003600a8f4149e43b97dac9ffafbbdf57c4"}]},"branch":"refs/heads/master"},"75949efb1d6af18ea971b9b84bf1ad52879cad41":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-07 03:52:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/788315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/15/788315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/15/788315/3"}}},"commit":{"parents":[{"commit":"e42724cc89493685934e9f99ef5f2061866b4280","subject":"Merge \"Add support for [cors] options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/e42724cc89493685934e9f99ef5f2061866b4280"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 15:21:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-07 03:52:15.000000000","tz":0},"subject":"Remove support for the Veritas HyperScale Driver","message":"Remove support for the Veritas HyperScale Driver\n\n... because it was deprecated during Victoria cycle[1] and driver\ncode was already removed from cinder.\n\n[1] 0da2b260eb6961222a23f7728c41593bcf349030\n\nChange-Id: I9af225f714bf2df6b4c579ff12392d5c30769cb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/75949efb1d6af18ea971b9b84bf1ad52879cad41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/75949efb1d6af18ea971b9b84bf1ad52879cad41"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
