)]}'
{"id":"openstack%2Fpuppet-cinder~792669","triplet_id":"openstack%2Fpuppet-cinder~stable%2Fwallaby~Ib913f0fe20af74c9a21875c6a7b497dfa2c30880","project":"openstack/puppet-cinder","branch":"stable/wallaby","topic":"tripleo-tempest","hashtags":["puppet-tripleo-job"],"change_id":"Ib913f0fe20af74c9a21875c6a7b497dfa2c30880","subject":"TripleO: Reduce tempest test scope","status":"MERGED","created":"2021-05-23 07:26:19.000000000","updated":"2021-06-16 14:08:41.000000000","submitted":"2021-06-16 14:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792669-tripleo-tempest","cherry_pick_of_change":791345,"cherry_pick_of_patch_set":1,"meta_rev_id":"589162b3741bebf0adda83610e7d326f6aac154d","_number":792669,"virtual_id_number":792669,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 14:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-16 02:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-05-25 16:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-16 02:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-23 09:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 16:06:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"77f530b74b9efad0be0df20be9d208acd6849266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-23 07:26:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"286df9b02bfaa6d3bdd4add0f640c0dfa3c9fc0e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-23 07:26:32.000000000","message":"Topic tripleo-tempest-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"05d143dad839ff8f961ec381107acf600bd8fb23","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-23 07:26:35.000000000","message":"Topic set to tripleo-tempest","accounts_in_message":[],"_revision_number":1},{"id":"430d07df86a15358ea973e60163823b3551a7dc9","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-23 07:28:28.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"eaff7595b06d1f9abb6c8c8c0c6d5081c1921eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 09:24:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d65cb2973c9d4716b5ef83c9c6bb7c39 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee98a11d358e44f2a3f8b25aae052c62 : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/424bd90016f3449fb936a151056d0bfc : SUCCESS in 12m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/631a5765509b4b859d6d090157355bdb : SUCCESS in 11m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbb2f3901cf9450da04a59ece566aa83 : SUCCESS in 11m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bf3da3a4332459a8cfeac9a44569f11 : FAILURE in 46m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d1d4d02fdcd48c8927699f1a8130f5a : FAILURE in 46m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e7c866a09174b43a0dd1c2f85337c73 : FAILURE in 32m 44s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4734ed2f9944ae7b8c4818e147f88ef : FAILURE in 31m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bea2db3fe448422f93847335a58e1535 : SUCCESS in 1h 29m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/68c7b7d09c444a50b9ff3ed9406b0c59 : SUCCESS in 1h 15m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f04571af66e147828f59b7f29b2098ef : SUCCESS in 1h 06m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/09c79877d3bc41669c5195f58d9b46b5 : SUCCESS in 1h 05m 19s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/428597121ee040cd800e91b68d4d95cf : SUCCESS in 21m 07s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8644ca13f754fe19bd823c2f90c94d4 : FAILURE in 15m 56s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bc6abcb4bb4f46b9a31449158f0eea62 : SUCCESS in 1h 56m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8f3e4124ec3fa616e4da6bdd644d58bae1e39aaa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-25 16:06:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a485e2a3690f30b8ff3f38488ec28d8912a7224","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 02:18:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI\u0027ll approve this now because this backport was already approved in the other repos...","accounts_in_message":[],"_revision_number":1},{"id":"4809e5fb9a27bc140807793de8374665295ce168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 02:19:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e75a7393f01429448512bde5a50e50f0d9ae2abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 03:54:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8770f79ccd614867b13b1225dbc89bf9 : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d8740fafa9b4708a6a3f1b521eb32e9 : SUCCESS in 4m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f290d58229d41aca6ea2806a6f6e1c0 : SUCCESS in 18m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/124d6f902e824b5683eabfaddd714862 : SUCCESS in 14m 30s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61dd4aef03fc4f37a65f4a0dd02d89fd : SUCCESS in 1h 04m 00s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95c074a275ed46c6a31fb9fc81e846db : SUCCESS in 1h 32m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad04b26b73a7487697e68256b997692b : FAILURE in 1h 29m 54s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b89d4f0fdc2c4c9d9af7653fe13e4434 : SUCCESS in 1h 13m 41s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3fb0e62af8f4bf099d424c0427c432b : SUCCESS in 23m 46s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c950abf994c4a9184e0a7ac7497c9fe : FAILURE in 16m 06s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/332ca1efc7014544a1297c295ab60103 : FAILURE in 1h 09m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2dd3cf4a5407bd882ff2349852db733b197ffbb9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 09:05:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc7b87627140c11135fa24fb9ec525373678c12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:42:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5752f9c7f8b541b99e9b3512ec90bf81 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abc888b9fec446f986905f2970756f53 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd8fee27fbfc4235833d22d1417e3b44 : SUCCESS in 22m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb786703022b4983b4f0adb04e7f6a45 : SUCCESS in 12m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/951dce1ff7ae4981b8c4e0196be21580 : SUCCESS in 12m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ae2c2edb13f40a7a518b221c7791fdb : FAILURE in 37m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f73d157a7654834b219a079703d682a : FAILURE in 45m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/875d37b87a38461086e6b5bc726eea5d : FAILURE in 30m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5771dbea13db41b1a78eeab3bffa6adc : FAILURE in 33m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/89fbd34526c64153a58ba0e9656f0924 : SUCCESS in 1h 15m 26s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b994c24e3534ec680ae3781cbafc5a3 : SUCCESS in 1h 27m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c286965584d4e01a5ad543ada737ba9 : SUCCESS in 1h 14m 27s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15c9e23b698640339fb25796e3462cda : SUCCESS in 58m 26s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/60908921fefd4d319277cccbc3297a0f : SUCCESS in 23m 27s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/013e611e9e2e4523a369c4404212a3c3 : FAILURE in 15m 28s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6db4913d448b4680a2e8cfe918355748 : SUCCESS in 2h 22m 37s","accounts_in_message":[],"_revision_number":1},{"id":"20b8df3bdb4d15c074b848daf678a20531786395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:43:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4575270025fdd55de68cba6247affc233b64cc6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:08:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64a47fa3917b4a0097f01eb732e49ec6 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad96c8305a5478f9769b2cc53ea06a6 : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/360c1009d58e4dbbba5f8405d0657c16 : SUCCESS in 15m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57cad91c1e504e4fb7eebdb448bb25eb : SUCCESS in 10m 57s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fb88aca214643679f470281b61d52c8 : SUCCESS in 1h 23m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4751110d06d74090995e6826a5208c5d : SUCCESS in 1h 38m 03s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/14a3927e3eac4e07bf5f1091616eea2f : SUCCESS in 1h 20m 56s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/662b076e44794a3e80de25592e8631c6 : SUCCESS in 1h 10m 47s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb82c3122c9c48d6a2d4943c5e7fcd50 : SUCCESS in 24m 09s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eb627b1a4e94337aa7fdae02955f8fa : FAILURE in 17m 51s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d018c794bbe94958940a93a516862d88 : SUCCESS in 2h 19m 41s","accounts_in_message":[],"_revision_number":1},{"id":"589162b3741bebf0adda83610e7d326f6aac154d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:08:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b2b7212e2388b199c28ca3b36e7d5be1bbd3e14","revisions":{"8b2b7212e2388b199c28ca3b36e7d5be1bbd3e14":{"kind":"REWORK","_number":1,"created":"2021-05-23 07:26:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/792669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/69/792669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/69/792669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/69/792669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/69/792669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/69/792669/1"}}},"commit":{"parents":[{"commit":"c7f4969e5060520690e3eb43f4d481639bc143e0","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/c7f4969e5060520690e3eb43f4d481639bc143e0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 02:30:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-23 07:26:19.000000000","tz":0},"subject":"TripleO: Reduce tempest test scope","message":"TripleO: Reduce tempest test scope\n\nThis change reduces scope of tempest tests in tripleo job, so that we\nrun only subset of test cases used in TripleO repos.\n\nChange-Id: Ib913f0fe20af74c9a21875c6a7b497dfa2c30880\n(cherry picked from commit 6bb2ecec0f0226fb7e9b57b37fc63ca4e544fdd3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8b2b7212e2388b199c28ca3b36e7d5be1bbd3e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8b2b7212e2388b199c28ca3b36e7d5be1bbd3e14"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
