)]}'
{"id":"openstack%2Fpuppet-cinder~793133","triplet_id":"openstack%2Fpuppet-cinder~master~I7c8c8b465ac47fe4609e9d8daa4848b132d383c7","project":"openstack/puppet-cinder","branch":"master","topic":"drop-fedora","hashtags":[],"change_id":"I7c8c8b465ac47fe4609e9d8daa4848b132d383c7","subject":"Drop Fedora support","status":"MERGED","created":"2021-05-26 09:53:51.000000000","updated":"2021-06-12 19:33:44.000000000","submitted":"2021-06-12 19:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793133-drop-fedora","meta_rev_id":"41034acb9ea2ce7451a1e849c4b0eefc5d38c6ce","_number":793133,"virtual_id_number":793133,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-12 19:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-06-01 14:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-06-12 19:28:08.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-06-12 19:28:08.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-26 10:14:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 15:51:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-12 19:28:08.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"fae75288d4b29090ea81e56349f149cea256db53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-26 09:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b08300464d2aadc429a7bcc4506e14d533efcbf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-26 09:55:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f9e01b1bfd484ef51cfe89a3520816e74da10610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 10:14:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/656786f4c9834d858d24879372a9c50f : SUCCESS in 2m 42s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b84dc98860a24829bacc1c8af30b6954 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76e96267be054fc389adaab57601c3e4 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4761f31d719b4b5aaca74d697befabd9 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d52b5f1962984dbf9e5786f607e0a2b9 : SUCCESS in 13m 19s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/c54a0786ce1648b5a2471d09e4648e77 : SUCCESS in 14m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad77ccd6f587400797b08419ee335129 : SUCCESS in 11m 17s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87f8423a1e8e43ef909f5bcda69881af : SUCCESS in 10m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f888c758109d4b538661ce3fc88b6c3e : SUCCESS in 9m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/234ad5fb8c374325af827f0612c6939b : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"97a6cce7b47a76a8c3a25121959d56b371deaf8d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-27 15:51:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"642830c3600f39bad7ccde178191c707deec7e89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-28 00:43:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6bec47b7af279559fa15965515dd925513b8c5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 00:51:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/08d1a018e9f24102aa54b1a2bb197519 : SUCCESS in 2m 59s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c5d2f40be024596948b5f9a83832fb4 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ccd5f277dd64e1fa7503041d4ae5d3f : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"aa28b074c11232acbc2de16ea434a246a04dbc53","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-01 14:51:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f49537a77b019fcedd54693d66200b47f260fc2d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-12 19:28:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"030e0435717fbb61c074b0345648cf8bfb42bea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 19:28:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ace059b04bee0287789e9d71128240ca45e1e89a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 19:32:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/28361838750a4a43b5667e1cbfce2827 : SUCCESS in 2m 31s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d799bc59525a4119bfa87e179eebbc41 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/211aabe57cef48238d067d8105cfe2cf : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"342447713d16da8284e6575b65c33dfb0982163e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 19:32:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"41034acb9ea2ce7451a1e849c4b0eefc5d38c6ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 19:33:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7839d8465598489fb841a4e216847273 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"27c6fe9700d9fa8324dc87008f75141682ceb8d8","revisions":{"ec9e8e2527b8525bd69a4f980e53ead24d3e5004":{"kind":"REWORK","_number":1,"created":"2021-05-26 09:53:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/793133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/33/793133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/1"}}},"commit":{"parents":[{"commit":"c90ca173686de83c9f61f04f0bd294b21c06ffac","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/c90ca173686de83c9f61f04f0bd294b21c06ffac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-26 09:53:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-26 09:53:38.000000000","tz":540},"subject":"Drop support for Fedora","message":"Drop support for Fedora\n\nSupport for Fedora is never tested, and has been unmaintained for\na while. Because we don\u0027t expect any actual user using OpenStack on\nFedora, this change drops support for Fedora directly.\n\nChange-Id: I7c8c8b465ac47fe4609e9d8daa4848b132d383c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ec9e8e2527b8525bd69a4f980e53ead24d3e5004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ec9e8e2527b8525bd69a4f980e53ead24d3e5004"}]},"branch":"refs/heads/master"},"becca294f2ace928ccfe65d0cb3ef83d0058f371":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-26 09:55:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/793133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/33/793133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/2"}}},"commit":{"parents":[{"commit":"c90ca173686de83c9f61f04f0bd294b21c06ffac","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/c90ca173686de83c9f61f04f0bd294b21c06ffac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-26 09:53:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-26 09:55:47.000000000","tz":540},"subject":"Drop Fedora support","message":"Drop Fedora support\n\nFedora support is never tested, and has been unmaintained for a while.\nBecause we don\u0027t expect any actual user using OpenStack on Fedora, this\nchange drops support for Fedora directly.\n\nChange-Id: I7c8c8b465ac47fe4609e9d8daa4848b132d383c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/becca294f2ace928ccfe65d0cb3ef83d0058f371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/becca294f2ace928ccfe65d0cb3ef83d0058f371"}]},"branch":"refs/heads/master"},"27c6fe9700d9fa8324dc87008f75141682ceb8d8":{"kind":"REWORK","_number":3,"created":"2021-05-28 00:43:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/793133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/33/793133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/33/793133/3"}}},"commit":{"parents":[{"commit":"a8123719c4ca05520aee90934cc4755a250a7e69","subject":"Merge \"Get rid of the $pyvers variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/a8123719c4ca05520aee90934cc4755a250a7e69"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-26 09:53:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-28 00:42:54.000000000","tz":540},"subject":"Drop Fedora support","message":"Drop Fedora support\n\nFedora support is never tested, and has been unmaintained for a while.\nBecause we don\u0027t expect any actual user using OpenStack on Fedora, this\nchange drops support for Fedora directly.\n\nChange-Id: I7c8c8b465ac47fe4609e9d8daa4848b132d383c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/27c6fe9700d9fa8324dc87008f75141682ceb8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/27c6fe9700d9fa8324dc87008f75141682ceb8d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
