)]}'
{"id":"openstack%2Fpuppet-cinder~799614","triplet_id":"openstack%2Fpuppet-cinder~master~Id4688cb7362d4be5161a8ea243448a2c269b2414","project":"openstack/puppet-cinder","branch":"master","hashtags":[],"change_id":"Id4688cb7362d4be5161a8ea243448a2c269b2414","subject":"Dropping explicit unicode literal","status":"ABANDONED","created":"2021-07-06 09:39:53.000000000","updated":"2021-08-27 03:53:11.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"60eba4f86a8771becd11c0efc04016387701edee","_number":799614,"virtual_id_number":799614,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-06 11:37:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":-1,"date":"2021-07-06 19:50:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 11:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 19:50:53.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-07-07 09:35:36.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-07-07 13:55:24.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-07 13:55:24.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"0b7a09d18367177068f07bedf2422e177e81008f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-06 09:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"552a47f75fd3108d9787837167dccfb2b0261e7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 11:37:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5d7c14417e143d2a281d66960a8c535 : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e52cfd142f5b465eae35cfa0e5bbc114 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b87a7e5efda241e6ba006ea5fc704767 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/601071880de34e56922c199fc0043554 : SUCCESS in 14m 01s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd535000ca694d5c9d4200b5e16378bd : SUCCESS in 15m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67fc5b6c625641b1816946c021869ee1 : SUCCESS in 10m 34s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90e454f6309b4581834ee4485f1ca750 : SUCCESS in 11m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f454d7854e2a40cb8fcc91fd1f229202 : SUCCESS in 10m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7e92704e67645969bd961a3109a93ba : FAILURE in 35m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdff7d06032c4819801e3901ef782409 : FAILURE in 42m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/068a63243b2e4123ada7c1675690a546 : FAILURE in 30m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9c193e5f2c34efa9e468988db8b56eb : FAILURE in 26m 42s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a6b7191eccf4e8e8bd51a1ccd98033c : SUCCESS in 23m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d504cbbbcb04a149eaf765b0d2f3185 : SUCCESS in 1h 28m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ad8516b98cc43ccb1caf34963b03428 : SUCCESS in 1h 27m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3972768a2b4241aa8fa011f689a0b040 : SUCCESS in 1h 25m 39s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/517d2e40d8354076a9f190340811f14f : SUCCESS in 1h 04m 10s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca6a85937faa459cb70f2ef76b2405a4 : SUCCESS in 23m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f07b94e22844633b96b09ec28cabdc2 : SUCCESS in 1h 14m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac054b283c864ebaa40e3514b6530dff : SUCCESS in 1h 29m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6bd11019750404c99b6ad026f32bb7e : SUCCESS in 1h 32m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2dadc576ac4e4a4492c071dfd7af0ce8 : SUCCESS in 1h 13m 19s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8082506276d4ebcb63e10c4a0646d20 : SUCCESS in 21m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d40e3516d684cd59b444906f53b7e68 : FAILURE in 14m 09s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b871e597b2544a9d84f144dce8e8dcd6 : SUCCESS in 21m 57s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd6c2c29cef44d9c9b5e4425d9214c15 : FAILURE in 14m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f045eead7e804392b3c3ab56bbd4935d : FAILURE in 14m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/427fd9e8891b40be91095adef7bc51ca : SUCCESS in 3m 47s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e2dca45f56844020954a352014023c6a : SUCCESS in 1h 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"173345cbecb6addc900f3eccd758e271bf54cc49","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-07-06 19:50:53.000000000","message":"Patch Set 1: Code-Review-1\n\nI agree with the changes you propose to conf.py, but I believe the ruby code should NOT be changed.\n\nThe reason is that while the \u0027u\u0027 prefix is no longer necessary with py3 (it\u0027s basically ignored), it is still syntactically valid. The cinder_type provider still needs to handle receiving an input with the prefix.\n\nI\u0027m curious if other reviewers agree with my assessment.","accounts_in_message":[],"_revision_number":1},{"id":"742b8b36541ebf61d6907fb7b62d083afad4cc44","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-11 05:04:01.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nRegarding the change in provider:\n\nAs per my quick testing python3 no longer put \"u\" in front of unicode\nBecause we rely on openstack client to get these outputs and now openstackclient no longer support Python 2.7, we no loner need to handle u\u0027...\u0027 and it\u0027d make sense to remove these implementation from ruby code as well.\n\nIn python 2.7\n~~~\n\u003e\u003e\u003e print({u\u0027foo\u0027: u\u0027baa\u0027})\n{u\u0027foo\u0027: u\u0027baa\u0027}\n~~~\n\nIn Python3.8\n~~~\n\u003e\u003e\u003e print({u\u0027foo\u0027: u\u0027baa\u0027})\n{\u0027foo\u0027: \u0027baa\u0027}\n~~~\n\n\nHowever, because the change in conf.py and one in provider code have a different characteristics, I\u0027d prefer creating a  separate patch for the ruby part.","accounts_in_message":[],"_revision_number":1},{"id":"4c0c27de3d73bf574ec8dbf05cdad4fe6a554a60","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-11 05:08:22.000000000","message":"Patch Set 1:\n\nInstead of handling u\u0027...\u0027, we might need to handle b\u0027...\u0027 in python3 because bytes type gets that \u0027b\u0027 as its prefix.\nHowever I\u0027m not very much sure whether bytes can appear in openstack volume type list...","accounts_in_message":[],"_revision_number":1},{"id":"60eba4f86a8771becd11c0efc04016387701edee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-08-27 03:53:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d5976c8716f11bbdd4c1c5ddb22e46610a6bf6a","revisions":{"8d5976c8716f11bbdd4c1c5ddb22e46610a6bf6a":{"kind":"REWORK","_number":1,"created":"2021-07-06 09:39:53.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/14/799614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/14/799614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/799614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/799614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/14/799614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/14/799614/1"}}},"commit":{"parents":[{"commit":"e42724cc89493685934e9f99ef5f2061866b4280","subject":"Merge \"Add support for [cors] options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/e42724cc89493685934e9f99ef5f2061866b4280"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 08:43:40.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 09:39:50.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: Id4688cb7362d4be5161a8ea243448a2c269b2414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8d5976c8716f11bbdd4c1c5ddb22e46610a6bf6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8d5976c8716f11bbdd4c1c5ddb22e46610a6bf6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
