)]}'
{"id":"openstack%2Fpuppet-cinder~822058","triplet_id":"openstack%2Fpuppet-cinder~master~Ibe59fbf893a7169ea6e3bf00a72b4df28b780b14","project":"openstack/puppet-cinder","branch":"master","topic":"cleanup-resource-type","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-23 10:36:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe59fbf893a7169ea6e3bf00a72b4df28b780b14","subject":"Deprecate redundant cinder::qos resource type","status":"MERGED","created":"2021-12-17 00:02:59.000000000","updated":"2021-12-23 10:38:21.000000000","submitted":"2021-12-23 10:36:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822058-cleanup-resource-type","meta_rev_id":"f2569d698c739ea83f1459066ba8f0ea5e8b0347","_number":822058,"virtual_id_number":822058,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-23 10:36:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-12-20 15:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-12-23 08:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-12-23 08:32:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-17 01:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 15:29:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-12-23 08:32:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d6108c1a039894113e4ab88bb28b84e3eba3d868","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-17 00:02:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68cb1112bd941e7568f9bdcf2560996fbde03bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-17 00:03:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5910261eeaab33e26770f95796c907471f55f95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 01:56:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15c994217f4d4e01ba49dea318438c00 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b7608dd9ff9498e86091f494208ac88 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98243da7f2bb42068b039002467d29ff : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/70e35ba3160a4dc6a06c126935b8299b : SUCCESS in 13m 54s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/55874cc59a41467daf5596926c4071c8 : SUCCESS in 12m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92a8e38abd2940eaad5640b5aba2a7a3 : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a239755c84a245d3b64480a9e0ab57d7 : SUCCESS in 11m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cf7a7d58e3f4920b312095f4a3f5016 : SUCCESS in 9m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e6fa7a4b3e444449499ed70a7ca1838 : FAILURE in 37m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46c047258afd4db6958f198b1c656b19 : FAILURE in 42m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e4e726dec58498a89e3335ec7c7a241 : FAILURE in 33m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b1a85fb0d7e4013886ad0748193a530 : FAILURE in 30m 37s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b5ec44828594d679e6d856f9a7feecc : SUCCESS in 27m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7e6f358f3ae402c917b40c2c9de771c : SUCCESS in 1h 28m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d575bb237d3746bd94d0b0af67040a1c : SUCCESS in 1h 40m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebd263ead9df4278abf0dc65404ea733 : SUCCESS in 1h 20m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa2b6584e317489a8aa7c63d4be3a77b : SUCCESS in 56m 48s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bb62858e94647ffb05dc1d61d43b2a4 : SUCCESS in 27m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c008302a4cd74917be8612de61fbe689 : SUCCESS in 45m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74874a6eefb24544b88c0a7699245288 : SUCCESS in 53m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df24a107f8f542ab99fbf758ce78efbb : SUCCESS in 59m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/62de07ea1b5845ae8c6aafcd41b3f46d : SUCCESS in 57m 18s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737f6f08acbd480283baa7020b60ad31 : SUCCESS in 17m 01s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6a0b87db1b24dc1bb0a4f313a8367be : SUCCESS in 21m 27s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/422e353e1cb240edae53194caf8102af : SUCCESS in 16m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e12309f74a7c4ca5a2a72416adafb9d8 : SUCCESS in 16m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/164df1f388404def811f83048615a655 : SUCCESS in 5m 01s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a39ba3c57d0542409490c14d3a1cb0a5 : FAILURE in 1h 50m 31s","accounts_in_message":[],"_revision_number":2},{"id":"43a684c098a75f55e6c39448ab741974bbe7048c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-17 05:30:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0648f91cc90618d95c488a8d44bc093443e2829f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 07:03:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b33c6b6255814fa389e396888021c243 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c552c20861724d79a69da711cb0d9b38 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70e96cf5a2d7479899a5850b617aaf26 : SUCCESS in 2m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/af7f5e0bccb3405dbefe7e1dd52f5ee0 : SUCCESS in 13m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6297294d072484f881f825394fef8c1 : SUCCESS in 12m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d740f310bce545bb97d82dd16bf455ae : SUCCESS in 12m 12s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4a3246d4b3e479eaeb870964e8bd8c2 : SUCCESS in 11m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3b260140cc644c0a0575279364b4a05 : SUCCESS in 10m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/001b249e6ede4a7d83f19be95e655dc8 : FAILURE in 29m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb48b80ab01a4974951be51195ca0bfe : FAILURE in 42m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d767e06fad6474c8f8c7e68d8482b58 : FAILURE in 37m 14s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bec353fec0b48008f54da0f458b0dcc : FAILURE in 24m 03s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fafe337e4b1349e194edf89e898c15b1 : SUCCESS in 21m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/726dc1e129a84fc6b0234b8c55508d3a : SUCCESS in 1h 15m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a10b4d98e29a4a21811fb59418989c1f : SUCCESS in 1h 28m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8079726a940b4eff8bead9b2092bd149 : SUCCESS in 1h 20m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bb8fe413fe5474c9820ecce6baae17c : SUCCESS in 1h 06m 13s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72750312b0ee4539a5940dc96736cc54 : SUCCESS in 22m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/81f42b57d24b4f6a81ea8b486ca5d3fc : SUCCESS in 50m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f27b611ede384889a0d7c95037c373ab : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc75621c91694b6aa653496878bf6165 : SUCCESS in 57m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2198d1832d7746c9abe860e15928ecaa : SUCCESS in 44m 47s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8438331558a74e448f5cb3a8d4b12f0a : SUCCESS in 15m 23s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a3e1a60baec4b209b18b365dbbfc3ed : SUCCESS in 22m 50s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b1b8ea2c7b140f29bddbef8e8870853 : SUCCESS in 16m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05095c5b5ff445e0bb075def4dfaa6f1 : SUCCESS in 16m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0758865209b4741aa8d31da5c8c996f : SUCCESS in 4m 36s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bd2dfcfb4c35465fb80fd2fb1629ed9a : FAILURE in 1h 26m 05s","accounts_in_message":[],"_revision_number":3},{"id":"76f589af80ca129dfc427b6c9abd9ce853397dea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 09:56:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"66cc0de1f7b4ab170d4d487d9c14790b78e878ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 12:11:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d54ef5483f3f487daa7974c31fb3942a : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fb45ea0a6d540718eb751dc2bd41ca5 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb4f8fd4a4fe4b4994d91fd94152f009 : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8f6812e76a54e17a07b68b95f1698a3 : SUCCESS in 15m 29s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a507d87d55454bbc93d3044382c746fa : SUCCESS in 15m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/631798e2283c49ed80ca1040c81137e6 : SUCCESS in 12m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f551158090f84700980109d84b1aa586 : SUCCESS in 11m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9a949f1af8244ebb3214637428b0fc8 : SUCCESS in 12m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce194d92150e4e858a19b69eab18799e : FAILURE in 33m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b28af270c2a4503bc08e79a5adf8ff6 : FAILURE in 35m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c248c634f23d430bb192041d999f2f0a : FAILURE in 33m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00b4979077ec437980af064bdcf04ed2 : FAILURE in 30m 00s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/547b08a092684dcdbc7014b4c52b9a1a : SUCCESS in 20m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fd14c89abc9470b9e57e05fc7bbb81f : SUCCESS in 1h 35m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a88ff0bb5e204f41a603b9680a83efa7 : SUCCESS in 1h 11m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8bbe48a0e57492c88f7b3e3f6ec047e : SUCCESS in 1h 19m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27fd6038d7664e04ae5416bcbf1c6a67 : SUCCESS in 1h 04m 47s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d13c36c0ab794bd89a6f9aa978a05797 : SUCCESS in 20m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c532cf9ed3db4377b06699318cabdccf : SUCCESS in 59m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc9576a7f92e4bf8852d2eadb15fcf6e : SUCCESS in 59m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac77586d83a34b3987614dde417b1dc0 : SUCCESS in 1h 06m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfca8895bd21448b9b46576ebdf0b3cd : SUCCESS in 46m 31s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e85e7fa68b38471eac41161547ae0a02 : SUCCESS in 16m 25s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba94536dd9c545308d75fb7ca43ae7b9 : SUCCESS in 25m 34s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/313b7fa369434a2f83e5d6d5b3d18f75 : SUCCESS in 16m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a59b8973a6e4dccba00fd7bc7dae087 : SUCCESS in 15m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d92a888a38f453fadf69856af0ae945 : SUCCESS in 4m 04s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d35e9c76357d4d07bcaa7e083efe1dec : SUCCESS in 2h 04m 14s","accounts_in_message":[],"_revision_number":3},{"id":"e4e3fac08515c0e57f5136d08eb52558b4a14e5a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-20 15:29:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"16c5d5faac3e1158a2920fccec4a9f8520299248","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-23 08:32:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ac61a12e8362ba6e91571eb8aa52e26f6364f126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 08:34:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5e22cfaf311df8d44e6a5b8f529f2950957030e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:36:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/148cd3adda454793b8e56f5f441cea5f : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6521aba44f384ff2bbd71a51890d153f : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e3e2b98dff74cf598bfdc391da54382 : SUCCESS in 4m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2046a77f7cfb4b868215db314e57cfb1 : SUCCESS in 15m 18s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff13d978a7f943aea07d380e35d42069 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9faa61b327d6436fa7423b7bb7aed112 : SUCCESS in 14m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29fbde1807e64c0ebc94125e1f90694f : SUCCESS in 11m 46s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48676ece1cfa4f97aa8363c08bac38e2 : SUCCESS in 1h 05m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/08a5a147b1334d62960cefad73be2fc0 : SUCCESS in 1h 29m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/40532fb8a5de427a999f16049cc9495a : SUCCESS in 1h 17m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0189e201edf244e5a3f86914883f3973 : SUCCESS in 55m 25s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d82fcf3334f4630abc573bf56b434eb : SUCCESS in 21m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f823c1f3e35424bbff949fb4327034a : SUCCESS in 19m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f055ac2f75f9443e91883ad55479b50d : SUCCESS in 3m 58s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b2a216559072400b848f1fac7afc7439 : SUCCESS in 2h 01m 29s","accounts_in_message":[],"_revision_number":3},{"id":"008e74a0c343aeaf75120914ca4bf454da86a976","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:36:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f2569d698c739ea83f1459066ba8f0ea5e8b0347","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:38:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52cee4fffb44bb08d03fe06025177d5 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0419e4a1f1542413bcb14d4668f3d9ca04f8f1c","revisions":{"21959a142f3ea809ebc125fb167e0893f1940877":{"kind":"REWORK","_number":1,"created":"2021-12-17 00:02:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/822058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/58/822058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/1"}}},"commit":{"parents":[{"commit":"03b5537d4f2dfe82df3e71bbab794e8dfa160faf","subject":"Merge \"Clean up direct dependencies on puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/03b5537d4f2dfe82df3e71bbab794e8dfa160faf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-17 00:01:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-17 00:01:56.000000000","tz":540},"subject":"Deprecate redundant resource typo for qos rule management","message":"Deprecate redundant resource typo for qos rule management\n\nThe cinder::qos resource type is just a wrapper of cinder_qos and\nhas no additional implementation. Let\u0027s deprecate it and let users\nuse the cinder_qos resource directly.\n\nChange-Id: Ibe59fbf893a7169ea6e3bf00a72b4df28b780b14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/21959a142f3ea809ebc125fb167e0893f1940877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/21959a142f3ea809ebc125fb167e0893f1940877"}]},"branch":"refs/heads/master"},"3f143ee29d98f4bd1e377f8b2f5faaac4d865e2e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-17 00:03:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/822058/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/58/822058/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/2"}}},"commit":{"parents":[{"commit":"03b5537d4f2dfe82df3e71bbab794e8dfa160faf","subject":"Merge \"Clean up direct dependencies on puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/03b5537d4f2dfe82df3e71bbab794e8dfa160faf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-17 00:01:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-17 00:03:18.000000000","tz":540},"subject":"Deprecate redundant cinder::qos resource type","message":"Deprecate redundant cinder::qos resource type\n\nThe cinder::qos resource type is just a wrapper of cinder_qos and\nhas no additional implementation. Let\u0027s deprecate it and let users\nuse the cinder_qos resource directly.\n\nChange-Id: Ibe59fbf893a7169ea6e3bf00a72b4df28b780b14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3f143ee29d98f4bd1e377f8b2f5faaac4d865e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3f143ee29d98f4bd1e377f8b2f5faaac4d865e2e"}]},"branch":"refs/heads/master"},"d0419e4a1f1542413bcb14d4668f3d9ca04f8f1c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-12-17 05:30:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/822058/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/58/822058/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/58/822058/3"}}},"commit":{"parents":[{"commit":"d944a76b95fb82fe161159954f3dfe720726bcfc","subject":"Deprecate service validation for cinder-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d944a76b95fb82fe161159954f3dfe720726bcfc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-17 00:01:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-17 05:30:28.000000000","tz":0},"subject":"Deprecate redundant cinder::qos resource type","message":"Deprecate redundant cinder::qos resource type\n\nThe cinder::qos resource type is just a wrapper of cinder_qos and\nhas no additional implementation. Let\u0027s deprecate it and let users\nuse the cinder_qos resource directly.\n\nChange-Id: Ibe59fbf893a7169ea6e3bf00a72b4df28b780b14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d0419e4a1f1542413bcb14d4668f3d9ca04f8f1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d0419e4a1f1542413bcb14d4668f3d9ca04f8f1c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
