)]}'
{"id":"openstack%2Fpuppet-cinder~854731","triplet_id":"openstack%2Fpuppet-cinder~master~I7c8cfb2cf7303fcb280e411276cf121bd1c9a74b","project":"openstack/puppet-cinder","branch":"master","topic":"apache-param-desc","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-26 15:55:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c8cfb2cf7303fcb280e411276cf121bd1c9a74b","subject":"Use standard parameter description format for wsgi::apache","status":"MERGED","created":"2022-08-26 06:16:09.000000000","updated":"2022-08-26 15:55:29.000000000","submitted":"2022-08-26 15:55:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"854731-apache-param-desc","meta_rev_id":"3754a328ab906df3e67af4f25bd2933d597844ca","_number":854731,"virtual_id_number":854731,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 15:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-26 09:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-26 14:05:32.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-26 14:05:32.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 07:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 09:05:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-26 14:05:32.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"c4663e3d489af34e940111c4a37eeb87680d3490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 06:16:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c6701a969cc18a7e06e7605e0707383d0a88eb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 07:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08fe5c22134b48f1b75b39f2651d9a37\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e2f7506904243f1acc5f9352863deaa : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9a972a82fb1496c86bf000737fcd529 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27ea5b93543246648b6681bf4d69200e : SUCCESS in 3m 23s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0c9fb83e8654f1884090e895865bc57 : SUCCESS in 17m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e0128fd67ce4f5b930555523cb530ae : SUCCESS in 12m 30s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a23ef8e49f9d4a0e9f669cb9db002c55 : SUCCESS in 12m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cca3b79ccace48b89b8218909bc9dcfd : SUCCESS in 12m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e32e0a30d7b403194a78f2f7a0974e4 : FAILURE in 59m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b87d7ecece0d48b4aafdde5a93bdad4e : FAILURE in 1h 00m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd4ead37a6294cafbf8a41db2ee82f2f : FAILURE in 44m 49s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f679adfb2974c2ebce5220814dd6972 : FAILURE in 42m 04s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c87a7bedd26b4be58a32953447d072e3 : FAILURE in 47m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd0c6bc70f4d4788bfd1af103ac00530 : SUCCESS in 24m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa5ebca4560e4652a103447f3be4e151 : SUCCESS in 59m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1dc2983ea774ae8bc1f83f7c45aa78d : SUCCESS in 1h 17m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f4f49e27e134c53ae568719095f2b87 : SUCCESS in 1h 21m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbaf6e6a7c47422191f44050364b4525 : SUCCESS in 1h 01m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de1cf4cc869b4808830e089177be807b : SUCCESS in 54m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9e027290c7c42ada46aadce1ff24e57 : SUCCESS in 19m 46s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/881e675cf84f4138869ae3c37d223206 : SUCCESS in 26m 06s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0b107a4e6764b0f84e5763c723dda09 : SUCCESS in 19m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83693e20f43c47c7850bbd58625af522 : SUCCESS in 15m 48s (non-voting)\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8e9a77590f5d4f33bd293c77356294e4 : SUCCESS in 1h 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"94f0ecea7572706a858eedeb3def161967016f88","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-26 09:05:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79fe44ebb2b9510ae30948ceefd0f2e1464e049d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-08-26 14:05:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1eb5ec19667525ecb5454e61b04252d7788d4653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 14:06:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20e14c3ca1cf44ade552388d7b96f334de9d9bec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 15:55:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/639fa545d0e84151ace6886cb057ed26\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/907e441197c64eb99353245fe6e380b2 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e9af7d7a0684fa4b350da359e647caa : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55a659f20d2e4d5b817a37a587497728 : SUCCESS in 3m 54s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/92a335ba0e4741babcc9989f4f08149e : SUCCESS in 18m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27b2c0e1fd074fdf82fae223df070332 : SUCCESS in 13m 13s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70279d9268d744cc9c2b9a22305a51b8 : SUCCESS in 13m 00s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbda375f9a874593b0cc38550039fa0b : SUCCESS in 1h 06m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9061fe60e3944df0944d4d78f794a60b : SUCCESS in 1h 14m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa0ef0939ab140188c77be288bde1d3b : SUCCESS in 1h 15m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7e26a7bbab14986bbcfa477d98483d9 : SUCCESS in 1h 02m 08s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e926587742949a8991ec0227bf81125 : SUCCESS in 1h 03m 36s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c36b008ed6684d3cb075f5e42e2ca07c : SUCCESS in 33m 21s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b09563741f7443189994e82b864b08c5 : SUCCESS in 1h 42m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3754a328ab906df3e67af4f25bd2933d597844ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 15:55:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cc758614af1ac5df32af3348dcc594ca0c7e759","revisions":{"8cc758614af1ac5df32af3348dcc594ca0c7e759":{"kind":"REWORK","_number":1,"created":"2022-08-26 06:16:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/854731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/31/854731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/31/854731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/31/854731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/31/854731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/31/854731/1"}}},"commit":{"parents":[{"commit":"6f7c66cf3e75a664ba4eb2cb47ee1785717e2342","subject":"Merge \"Support [DEFAULT] image_conversion_disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/6f7c66cf3e75a664ba4eb2cb47ee1785717e2342"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 06:15:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 06:15:45.000000000","tz":540},"subject":"Use standard parameter description format for wsgi::apache","message":"Use standard parameter description format for wsgi::apache\n\nCurrently parameter description of the \u003cmodule\u003e::wsgi::apache classes\nare formatted differently in individual modules, and this is making\nthe maintenance effort quite difficult.\n\nThis change updates the description format following the srandard one\nwe are globally using in our modules to reduce undesired differences\nbetween modules.\n\nChange-Id: I7c8cfb2cf7303fcb280e411276cf121bd1c9a74b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8cc758614af1ac5df32af3348dcc594ca0c7e759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/8cc758614af1ac5df32af3348dcc594ca0c7e759"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[]}
