)]}'
{"id":"openstack%2Fpuppet-cinder~863212","triplet_id":"openstack%2Fpuppet-cinder~stable%2Fxena~I7743344c150d51c0db5e7d6f6d8ba509cc1db7e1","project":"openstack/puppet-cinder","branch":"stable/xena","topic":"lp/1966382-stable/xena","hashtags":[],"change_id":"I7743344c150d51c0db5e7d6f6d8ba509cc1db7e1","subject":"setuptools: Disable auto discovery","status":"MERGED","created":"2022-11-02 09:03:58.000000000","updated":"2022-11-02 10:52:45.000000000","submitted":"2022-11-02 10:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863212-lp/1966382-stable/xena","cherry_pick_of_change":835560,"cherry_pick_of_patch_set":1,"meta_rev_id":"eb29fb154dc2a1ed8c77e41f45436a7c30109470","_number":863212,"virtual_id_number":863212,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-02 10:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-11-02 10:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-11-02 10:48:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 09:03:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"CC"},{"updated":"2022-11-02 09:11:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3aa2ae31feab2f488db00704adbe58768cce1f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 09:03:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bc60c175ae9c5c9002b618550996509d8eeb6ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 09:11:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1233779eedf14efc8b5d057af529084b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec6388f0d03c4992925d115a01a48a60 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"09fb565939bad35296bec9051db8be27c6639bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 09:14:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f8f19132d1c16b1dafffc5fc4062a06f813f530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 09:18:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a617e6fbe06c48df800faba605167a4f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4da47d1787f4fe4871fe96780f4e45b : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ca852a7faf788817686583991d2c044fb07fa342","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 10:48:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"298fde888cd6a13607a260d012ba5758f70d60af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 10:48:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a5151c9067754bfa376ebd42d527f79806ed138f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 10:52:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f06390c1914e84818316aaefac7460\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db1b953fa32244c8bfbf666bcb4830e6 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"eb29fb154dc2a1ed8c77e41f45436a7c30109470","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 10:52:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ddf2a6f74aee04f706db46064e55c2b4635dde74","revisions":{"b44e8f2877d8878c8743ddab6e3823fc6b45f0b2":{"kind":"REWORK","_number":1,"created":"2022-11-02 09:03:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/863212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/12/863212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/1"}}},"commit":{"parents":[{"commit":"605430e6c437681f46c4dd309d84266b83fbb99c","subject":"lvm: Modify targets.conf in Ubuntu/Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/605430e6c437681f46c4dd309d84266b83fbb99c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:37:35.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 09:03:58.000000000","tz":0},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I7743344c150d51c0db5e7d6f6d8ba509cc1db7e1\n(cherry picked from commit d0d981b6de20a8374f28a391b26e44b98d344a63)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b44e8f2877d8878c8743ddab6e3823fc6b45f0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b44e8f2877d8878c8743ddab6e3823fc6b45f0b2"}]},"branch":"refs/heads/stable/xena"},"ddf2a6f74aee04f706db46064e55c2b4635dde74":{"kind":"NO_CHANGE","_number":2,"created":"2022-11-02 09:14:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/863212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/12/863212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/12/863212/2"}}},"commit":{"parents":[{"commit":"605430e6c437681f46c4dd309d84266b83fbb99c","subject":"lvm: Modify targets.conf in Ubuntu/Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/605430e6c437681f46c4dd309d84266b83fbb99c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:37:35.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 09:14:15.000000000","tz":0},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I7743344c150d51c0db5e7d6f6d8ba509cc1db7e1\n(cherry picked from commit d0d981b6de20a8374f28a391b26e44b98d344a63)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ddf2a6f74aee04f706db46064e55c2b4635dde74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ddf2a6f74aee04f706db46064e55c2b4635dde74"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
