)]}'
{"id":"openstack%2Fpuppet-cinder~865794","triplet_id":"openstack%2Fpuppet-cinder~master~I9ebb3f6a214222457547f687c6f3abd2b9c058fb","project":"openstack/puppet-cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-29 19:08:16.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-11-28 17:43:18.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}},"hashtags":[],"change_id":"I9ebb3f6a214222457547f687c6f3abd2b9c058fb","subject":"NetApp: Deprecate support for netapp_copyoffload_tool_path","status":"MERGED","created":"2022-11-28 06:48:24.000000000","updated":"2022-11-29 19:09:19.000000000","submitted":"2022-11-29 19:08:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865794","meta_rev_id":"bfb2090c838798c2f8fffa6067053fbc3dc69a04","_number":865794,"virtual_id_number":865794,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 19:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2022-11-28 17:43:18.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2022-11-29 16:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2022-11-29 16:58:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 08:31:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 15:49:43.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2022-11-28 17:43:18.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-11-29 16:58:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"9b3e756a0d87287570193b517a1d7e616435b019","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 06:48:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3526e69d2faccf0a22147feb42007d3288f135d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 08:31:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f2bade91efe41738d54d5b4ee3c45df\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccbedcfa24ed426f8fec9fec58a9194a : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14fe03d4f1424400b29590b85b66e148 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62489a5ca8134b6b835fde2dde90c877 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c590bf4030774145986da70cebf2ed04 : SUCCESS in 9m 56s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/923406e5c1214de2b909622206c79403 : SUCCESS in 8m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45b65fae3bed4d76a09884d41e7260a9 : SUCCESS in 9m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c7f070bc2804200aa761c879144e126 : SUCCESS in 9m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2256e77c9bcc46b485bdb6f075516c19 : FAILURE in 20m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f4d38dd83a14b8d8aebb52b285e359c : FAILURE in 17m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9475f9ee062a4e96b411561921f0fb0e : FAILURE in 18m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/268f2a6530e043c68cbc73a09fef6be1 : FAILURE in 16m 51s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc6c0dbea9b441ccbde2c9982683c6fe : FAILURE in 16m 17s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9289a0f916f2400d873df347d55d39ff : SUCCESS in 20m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc2e0013986d4961b935f0fcedbd7175 : SUCCESS in 59m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a13727097094c07899c02d9bc52c0d6 : SUCCESS in 1h 13m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36a403d8469d4004a56dc1f21e631429 : FAILURE in 1h 31m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2663a98f30324f88b0846e14b2135c53 : SUCCESS in 55m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/959c4b2a30ae4f3cb168148150c70e42 : SUCCESS in 40m 40s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90553ec8051e484b8622f5d4e9cc3993 : FAILURE in 10m 25s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/335190593359481bac8b9bd6a2e29676 : SUCCESS in 19m 43s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/210aaf6ae8c44565a70e716368ac7724 : FAILURE in 25m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/607b9f2ecfde45dd8b669525e694d7e3 : FAILURE in 25m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bca33929aca470f8cca0cb5068c735f : SUCCESS in 5m 50s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8da4ed66d4ee4369831b55c49d26d9c7 : SUCCESS in 1h 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fe9a41027451404bd16518186e563b4c8d62e70d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 08:33:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"313b57121930c2d9415f7ad45857fea77d9a8062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 10:20:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7de932c65cb141a68ee01eed9a392e2f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91d35bb8ba9c4195bef22313bfbf7bd2 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e079b03a7c2c41949908078e95acbf1e : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db3ff8df9e1d4662ba236139c75933a6 : SUCCESS in 2m 33s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/516fd5903aa4404e880cc1276e8869b3 : SUCCESS in 10m 12s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27cc62713f7845d1ace0c61887b93dcc : SUCCESS in 7m 37s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4354aae62ca43d2a8405a7c7073b4a0 : SUCCESS in 10m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ad3ab68b24644bb8aad05ee274df77a : SUCCESS in 7m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/682b08950b3147d791bac8b8a29162b9 : FAILURE in 17m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f48feb21a96c4e54a6de001ff4bf0ea1 : FAILURE in 15m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aecc288e797045faaec36ba2cd78d84e : FAILURE in 15m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc365d0a2c254032b694090cb824498c : FAILURE in 17m 03s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7542dbffab24d16aae546411292864a : FAILURE in 12m 45s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b91fa3192a384f968aef192b24a5cafb : SUCCESS in 20m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d0b3910e83540fd87e14274ecb0a841 : SUCCESS in 59m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7960f56866b4e099b7b057d22807fc1 : SUCCESS in 1h 02m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6489d426b7894f4a9fdff7d5761ffef0 : SUCCESS in 1h 00m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/285d40f8239042e3bfb7bd09db4aa319 : FAILURE in 49m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bf0ae609f8e40908ac59461aa6cbac1 : SUCCESS in 47m 32s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2a145143c20471c89ce98821ba21e10 : FAILURE in 12m 06s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/95c9683445494e9a8068e4efb256935c : SUCCESS in 16m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa37028244e04c6989fb26cfc5b17dbc : FAILURE in 26m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06692e5232e44eec82ab00e610d7d55a : FAILURE in 23m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6512ed3070894527bba0954d4e15da8b : SUCCESS in 2m 54s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1fa5e7fe442b495ba3a1f06c9a0531fb : SUCCESS in 1h 42m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c79dd0f23eca7c25c119f93c1f41ba8fb34223b8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 13:46:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c06a3ba55669b54d7b7b6543f5554f71d560a581","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-11-28 15:49:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4830d7b4a49a6f6da13f54736491b570b17816c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 15:54:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8ab90f7d1a9f914c8577913d6dcef838b75af98c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 15:55:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94049917d4ba75b1dd52d4836876470943634497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 17:42:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9421a804ea3f4e34a790195cd8269242\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e52c8668e74446839c9e2ae0127c27fd : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c6c71a508c04ae7a4f7ad13d1c02ca0 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b93f3e4399864b318567a1827e4e96f7 : SUCCESS in 4m 01s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/dba129e10a6d49889053acbdf71bbe5f : SUCCESS in 15m 29s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c1edfceddbb4f7194bfef6b408d6683 : SUCCESS in 10m 43s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dcca3d7d5fb4fbd8fa8eb68439840fe : SUCCESS in 10m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f145db193d094c249b4b60cdaa528319 : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d5d5ecfa28346de8672e53d4323a034 : FAILURE in 18m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9013affcef1242c69ad433e9678c2f16 : FAILURE in 16m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/043c708709d24a859747dc19a2fdbda9 : FAILURE in 18m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96884bbd70714d87b3e7e82bd3e96420 : FAILURE in 17m 51s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/153dac6072d145de8c51420604f9cc2c : FAILURE in 15m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/566d70b2233549c0ba3a8fbec72cf8eb : SUCCESS in 21m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f3747a6cfad4a7a9a3555d0371ca706 : SUCCESS in 1h 02m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/98a51022d68847b8b315403ba5b5bbe7 : SUCCESS in 1h 12m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a357fdb46594cacaf05e08383909e99 : SUCCESS in 1h 10m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/06d029e8cc114c8486d9c68b3fe59b3d : SUCCESS in 56m 09s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab3ab3a616554a55a6dd40348d089993 : SUCCESS in 56m 19s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3271d0ba5bc441f79a520840da18120b : FAILURE in 11m 23s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e5496d4eb2b4cbab3be7553e40909cd : SUCCESS in 22m 35s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0b16462eb14b748438f996d2e12397 : FAILURE in 27m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/639cd90135db4e7a8e9e963f31e1dbfe : FAILURE in 25m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fed85db474f741a084c4806cdf759259 : SUCCESS in 3m 26s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ce7a0e3e22a244dea8ea70ee92c4922b : SUCCESS in 1h 45m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ed40b93559ffc949b44ea3177353e287348a0914","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-11-28 17:43:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"247baf0c773a15dcac42839f6657172d744ad735","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-29 16:58:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d063e3c0f752e06e3bd757f681db9c205d7348c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 16:59:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d428e23f44c0ba45914ae7ecb2a412b7644d477","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:08:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59337164ebdb42dd8d5e4da3e7af88d0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfb6f52f87c54a15a12cc26dc65894c8 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ac968d288304c92a916bc49154658e9 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f672c760ba094ce3ad9d3227db483263 : SUCCESS in 3m 50s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9610734fbd347639c8c866e71519204 : SUCCESS in 14m 26s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32257268390447b182aa6540e64e9e16 : SUCCESS in 9m 32s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92ce4625857479d890012d671c2a418 : SUCCESS in 8m 21s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb13aec82b0349d68d1b97c244ea26bd : SUCCESS in 1h 00m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23387f2f496441ecacc5d4814a31674b : SUCCESS in 1h 14m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8bb574810be43cfadada3e0beeda69d : SUCCESS in 1h 01m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71c9df8587a04e38a8104d12b6df3d05 : SUCCESS in 54m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49e919b4377040c5ac84dfbd57920baa : SUCCESS in 55m 42s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/21b2bc7ba0c84cfb903306e9dc902efe : SUCCESS in 22m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b43e194ef0df40caad3545a81d445c9d : SUCCESS in 3m 04s\n- puppet-cinder-tripleo-standalone https://zuul.opendev.org/t/openstack/build/03cc46a35e474637bf3957bc109a34a2 : SUCCESS in 2h 02m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4cb3d085fb2de14a85c955abb47fb5f7f5f2590e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:08:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bfb2090c838798c2f8fffa6067053fbc3dc69a04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:09:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40cea528c8434bd48b78a8cade859b47\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93bcdba1df88428c84de47087ee49e6c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"daa5d5cb188daaf3ecd9a9a97f087416feca26ae","revisions":{"76fe0e5dedd32a090657f597f90f492e99ef4fb8":{"kind":"REWORK","_number":1,"created":"2022-11-28 06:48:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/94/865794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/1"}}},"commit":{"parents":[{"commit":"3a63b7145b8a98e3a1189bae98a6b7a7834c2f6d","subject":"NFS driver: Remove deprecated nfs_disk_util","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3a63b7145b8a98e3a1189bae98a6b7a7834c2f6d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 06:47:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 06:48:14.000000000","tz":540},"subject":"NetApp: Deprecate support for netapp_copyoffload_tool_path","message":"NetApp: Deprecate support for netapp_copyoffload_tool_path\n\n... because the parameter was already deprecated in cinder.\n\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/847733\nChange-Id: I9ebb3f6a214222457547f687c6f3abd2b9c058fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/76fe0e5dedd32a090657f597f90f492e99ef4fb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/76fe0e5dedd32a090657f597f90f492e99ef4fb8"}]},"branch":"refs/heads/master"},"daa5d5cb188daaf3ecd9a9a97f087416feca26ae":{"kind":"REWORK","_number":2,"created":"2022-11-28 15:54:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865794/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/94/865794/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/94/865794/2"}}},"commit":{"parents":[{"commit":"3a63b7145b8a98e3a1189bae98a6b7a7834c2f6d","subject":"NFS driver: Remove deprecated nfs_disk_util","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3a63b7145b8a98e3a1189bae98a6b7a7834c2f6d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 06:47:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 15:54:18.000000000","tz":0},"subject":"NetApp: Deprecate support for netapp_copyoffload_tool_path","message":"NetApp: Deprecate support for netapp_copyoffload_tool_path\n\n... because the parameter was already deprecated in cinder.\n\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/847733\nChange-Id: I9ebb3f6a214222457547f687c6f3abd2b9c058fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/daa5d5cb188daaf3ecd9a9a97f087416feca26ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/daa5d5cb188daaf3ecd9a9a97f087416feca26ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
