)]}'
{"id":"openstack%2Fpuppet-cinder~901105","triplet_id":"openstack%2Fpuppet-cinder~master~I21fc46f49a143ae1282cad91732925195eda6541","project":"openstack/puppet-cinder","branch":"master","topic":"provider-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-17 18:40:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21fc46f49a143ae1282cad91732925195eda6541","subject":"Stop reading keystone_authtoken options","status":"MERGED","created":"2023-11-16 06:14:42.000000000","updated":"2023-11-17 18:41:05.000000000","submitted":"2023-11-17 18:40:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901105-provider-cleanup","meta_rev_id":"2e3e26f603ba7f99abb39ff88196f7b23174d2c5","_number":901105,"virtual_id_number":901105,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 18:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-17 14:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-17 15:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-17 15:10:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 08:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 14:08:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-17 15:10:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"45794da0afb8b04f7ad38351218de314022207e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:14:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15ba0f1ea37b11cd534e21e62d324a3b97a7d93b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:15:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8819b5bf0260dacb36327204d6ffee59536dad0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:17:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"600c362909e56e11350d40b5aefa9d9abc3c74df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:18:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"60b81f2ed006e013a445046d906f0baf22ba99af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:32:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7b48e806e31cd4eb708e15ca5302ed4301762908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 07:01:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"80ce28a53ef4e61d3614469e3b42a51e0f529338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 08:13:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb672cdbe49b4ea3b1d2e55dd0211362\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/356e2c82cf754173a468f177face6493 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d7b9bed09a64248ad840f0e7d8150df : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8611bf5a61db487a948b33dc6e54215b : SUCCESS in 5m 26s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e74684cc0c24a4b8626033a696acc21 : FAILURE in 13m 45s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5b36d06f17a41fa9d8a99e143828421 : FAILURE in 9m 02s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cd8e7efb696485984b116d3aa9e25e2 : FAILURE in 12m 25s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/650754eab270405ca25c33e89e7a3ad9 : FAILURE in 9m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8de8555d27144958f01d6cbeb9aea20 : SUCCESS in 45m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e82373e048447dabf49c0cf3181db69 : FAILURE in 52m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1905ebfbe2d34391a67c5e92a71ebe67 : SUCCESS in 49m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8426daf1c6344228bb16d106fbab7ddb : SUCCESS in 39m 49s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f0ca054168041e4998d517b8d02f3f6 : SUCCESS in 41m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5630ba86c0d64ee4a6b0d71a5d540725 : SUCCESS in 1h 03m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc3fd3d6466b4d67b4717376607ffb98 : SUCCESS in 1h 08m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c41706b66d24d52a970efc42d863fa1 : SUCCESS in 53m 40s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d690bb493304e479977d142b48a889a : SUCCESS in 52m 23s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1347f363aba4f2aa8ddc26065903bc1 : SUCCESS in 46m 10s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbbd84bb2cc9483b864839fea641031e : SUCCESS in 20m 29s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50b4ffff903b4d1ca916155fbeb2eb1b : SUCCESS in 14m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7182511fbc4d428f95dec339d00e45d5 : SUCCESS in 14m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/702ec6261fec471c90909df3cac16681 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":6},{"id":"3a5ad8b411b564ee45ea2df4dd48b7a0f66edac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 08:36:09.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8b2ac38dc480bd0a2456941c6d74fb307ccd2cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 10:01:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2b73a15ba6d4aa6b59e8f8295df7453\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7345319ac6a34218bd0e40752ab06f66 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/191134fdec08417eb8dcd4f51873b3d5 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbcc04ecc521425d90f99c35fd64f4be : SUCCESS in 5m 58s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/5320f00a154b4a35a17c2f1055883d00 : SUCCESS in 13m 15s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a669210a3a64572a19e59b94029f66b : SUCCESS in 10m 26s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/02f020c98a2b4b16bb4ca0b439f28bc4 : SUCCESS in 13m 35s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a792935b58924aa1981a8c8162601f5a : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72e7f6ff62294fd885ea982c1f0bffb9 : SUCCESS in 43m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8888a41293cb4e05a551fdb1349a5c4c : SUCCESS in 56m 23s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/015a0d34e6ff4aa78b7c7437b630b2e4 : SUCCESS in 57m 52s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c1e16d54c334f89a26fb66d1404690f : SUCCESS in 43m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/841844f8a04240889282a3f648b010ea : SUCCESS in 41m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/69b92d000ec54899939e403323689309 : SUCCESS in 57m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/acadda4367c341968b4a6ad7d5c3bf08 : SUCCESS in 1h 17m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2aa73eef91484e68ab27e628cb9c4288 : SUCCESS in 1h 01m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/244971df6b444db3ba29dd0339f962cf : SUCCESS in 52m 21s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58bc92b6147144d6ad73f3d6688100fd : SUCCESS in 52m 00s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d200334837f4663b24f5ec67bdecc11 : SUCCESS in 19m 07s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7b7a300884f4d9ca3b7ab0e8a9bdde7 : SUCCESS in 14m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11ffbfc922d64d808e10711f14bf4773 : SUCCESS in 14m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af665dc7d58a4b56bb30056ec7378ce1 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":7},{"id":"070ec26e3069039e6c1274665ed8149e1a00c855","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-17 14:08:14.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0c8591d593d757882d5cbc81c7f53a3e2e753aaf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-17 15:10:34.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"351db9c62d930aca3982a9be875bc1b1357a011e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 15:11:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3f30530c0a2b4aebbafa1a35e2e1e18e814cb8f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:40:03.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55280d673d144f39af7307c1b006129d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/235129a311d9498b8e676f3920d1e236 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a293f77d9aef49a0b2a8c772be799bc2 : SUCCESS in 3m 34s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ef913341bde436597f14873a757d919 : SUCCESS in 10m 27s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5992275cfd064db5baa9ba7509acb917 : SUCCESS in 9m 41s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e00063e6f8c4ed089e3f435b942434c : SUCCESS in 12m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d54a378e8c1547ec9f41aa2705101886 : SUCCESS in 1h 00m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fcb537a9d29408ab29350fa32f3f567 : SUCCESS in 1h 08m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1982c1e3d84e4fb49c9aaed64f8887b0 : SUCCESS in 1h 04m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b906526691bc4bcda4312604d6407049 : SUCCESS in 1h 08m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/709b1a8caa1140dcb65edb7f809c7947 : SUCCESS in 1h 00m 57s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/90a45bc126c94f8c95c990f15b6813df : SUCCESS in 17m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4c23673d9aa445d9f64a5dd150eb61a : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":7},{"id":"92b7137797060d3198df84cb28a349e5686dea7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:40:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"2e3e26f603ba7f99abb39ff88196f7b23174d2c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:41:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac2a34ed24b647b08a992a273cece332\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/080f9342f770455fb18d49a3d5211349 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"04e3d15f1a57fc9593a3e94d4e02714afa5f4a56","revisions":{"73538cff0e35a6ec8d402867f4a9a6c3357aaf78":{"kind":"REWORK","_number":1,"created":"2023-11-16 06:14:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/901105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/05/901105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/1"}}},"commit":{"parents":[{"commit":"5451b7135b486e8d7a8936adbc0f0d5298f5c3c3","subject":"Add support for reserved_percentage option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5451b7135b486e8d7a8936adbc0f0d5298f5c3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:26.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for cinder_* resources\nwas deprecated some cycles ago[1].\n\n[1] 596a77483b3d2c25590ab58b71037de9c4d1c81d\n\nChange-Id: I21fc46f49a143ae1282cad91732925195eda6541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/73538cff0e35a6ec8d402867f4a9a6c3357aaf78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/73538cff0e35a6ec8d402867f4a9a6c3357aaf78"}]},"branch":"refs/heads/master"},"d97e2b327ed07f72bd34997b0e1080d313019c6e":{"kind":"REWORK","_number":2,"created":"2023-11-16 06:15:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/901105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/05/901105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/2"}}},"commit":{"parents":[{"commit":"5451b7135b486e8d7a8936adbc0f0d5298f5c3c3","subject":"Add support for reserved_percentage option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5451b7135b486e8d7a8936adbc0f0d5298f5c3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:15:09.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for cinder_* resources\nwas deprecated some cycles ago[1].\n\n[1] 596a77483b3d2c25590ab58b71037de9c4d1c81d\n\nChange-Id: I21fc46f49a143ae1282cad91732925195eda6541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d97e2b327ed07f72bd34997b0e1080d313019c6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d97e2b327ed07f72bd34997b0e1080d313019c6e"}]},"branch":"refs/heads/master"},"459ac989483294dbaa35b3d3cc75dc585825b0cf":{"kind":"REWORK","_number":3,"created":"2023-11-16 06:17:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/901105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/05/901105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/3"}}},"commit":{"parents":[{"commit":"5451b7135b486e8d7a8936adbc0f0d5298f5c3c3","subject":"Add support for reserved_percentage option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5451b7135b486e8d7a8936adbc0f0d5298f5c3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:16:58.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for cinder_* resources\nwas deprecated some cycles ago[1].\n\n[1] 596a77483b3d2c25590ab58b71037de9c4d1c81d\n\nChange-Id: I21fc46f49a143ae1282cad91732925195eda6541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/459ac989483294dbaa35b3d3cc75dc585825b0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/459ac989483294dbaa35b3d3cc75dc585825b0cf"}]},"branch":"refs/heads/master"},"add72e9fb597f9e3abc040d30da73214db0e76b0":{"kind":"REWORK","_number":4,"created":"2023-11-16 06:18:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/901105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/05/901105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/4"}}},"commit":{"parents":[{"commit":"5451b7135b486e8d7a8936adbc0f0d5298f5c3c3","subject":"Add support for reserved_percentage option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5451b7135b486e8d7a8936adbc0f0d5298f5c3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:18:45.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for cinder_* resources\nwas deprecated some cycles ago[1].\n\n[1] 596a77483b3d2c25590ab58b71037de9c4d1c81d\n\nChange-Id: I21fc46f49a143ae1282cad91732925195eda6541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/add72e9fb597f9e3abc040d30da73214db0e76b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/add72e9fb597f9e3abc040d30da73214db0e76b0"}]},"branch":"refs/heads/master"},"b8714790b569399cc6428282f15c68bc7def1bdf":{"kind":"REWORK","_number":5,"created":"2023-11-16 06:32:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/901105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/05/901105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/5"}}},"commit":{"parents":[{"commit":"5451b7135b486e8d7a8936adbc0f0d5298f5c3c3","subject":"Add support for reserved_percentage option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5451b7135b486e8d7a8936adbc0f0d5298f5c3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:32:32.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for cinder_* resources\nwas deprecated some cycles ago[1].\n\n[1] 596a77483b3d2c25590ab58b71037de9c4d1c81d\n\nChange-Id: I21fc46f49a143ae1282cad91732925195eda6541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b8714790b569399cc6428282f15c68bc7def1bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b8714790b569399cc6428282f15c68bc7def1bdf"}]},"branch":"refs/heads/master"},"5c0c2ad4851922b27351cfee126aaeca5350926a":{"kind":"REWORK","_number":6,"created":"2023-11-16 07:01:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/901105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/05/901105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/6"}}},"commit":{"parents":[{"commit":"5451b7135b486e8d7a8936adbc0f0d5298f5c3c3","subject":"Add support for reserved_percentage option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5451b7135b486e8d7a8936adbc0f0d5298f5c3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 07:01:45.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for cinder_* resources\nwas deprecated some cycles ago[1].\n\n[1] 596a77483b3d2c25590ab58b71037de9c4d1c81d\n\nChange-Id: I21fc46f49a143ae1282cad91732925195eda6541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5c0c2ad4851922b27351cfee126aaeca5350926a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5c0c2ad4851922b27351cfee126aaeca5350926a"}]},"branch":"refs/heads/master"},"04e3d15f1a57fc9593a3e94d4e02714afa5f4a56":{"kind":"REWORK","_number":7,"created":"2023-11-16 08:36:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/901105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/05/901105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/05/901105/7"}}},"commit":{"parents":[{"commit":"5451b7135b486e8d7a8936adbc0f0d5298f5c3c3","subject":"Add support for reserved_percentage option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/5451b7135b486e8d7a8936adbc0f0d5298f5c3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 08:36:01.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for cinder_* resources\nwas deprecated some cycles ago[1].\n\n[1] 596a77483b3d2c25590ab58b71037de9c4d1c81d\n\nChange-Id: I21fc46f49a143ae1282cad91732925195eda6541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/04e3d15f1a57fc9593a3e94d4e02714afa5f4a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/04e3d15f1a57fc9593a3e94d4e02714afa5f4a56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
