)]}'
{"id":"openstack%2Fpuppet-cinder~904318","triplet_id":"openstack%2Fpuppet-cinder~master~Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db","project":"openstack/puppet-cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-26 08:24:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db","subject":"cinder_type: Accept hash for properties","status":"ABANDONED","created":"2023-12-25 05:47:21.000000000","updated":"2024-10-07 15:50:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8646d998241291e3811764097ee2b0dc0b0c38bf","_number":904318,"virtual_id_number":904318,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-26 04:12:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-12-26 08:24:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-25 08:52:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92b7db7f91e7822707f91030ff72cc79a1c5c2a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 05:47:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2c04b897132f08b7d23725f6db35acf3ef58b85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 06:46:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"821970c70926d46e579f9a66c895353782f3ce42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 06:56:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5ead818e1c7155bf82d4ec23f4d4b2fd3f93e493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 07:07:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"35cb1c78b81fb8922cfc46930d89aeeed0b0687a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 07:41:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"28424c14c1c278259b0e42cf960d047fad648bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 08:52:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3299b338f3384423b2a7e3b4b1d1d0f7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a1a2aeb450249b9b9d7545d523c1ddb : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b37621f85ffd4a95910cf5638ac1fc61 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c389a82529fa447b9a3e86ced9b92800 : SUCCESS in 5m 34s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/19256e40a6034349b1954fb2db2b2f00 : FAILURE in 12m 43s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7f354d765f44279b37d21f18524fe31 : FAILURE in 8m 53s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/c73cebb2f3ec4c93a56c4bb8b40ae230 : FAILURE in 12m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/527bbcf1cdbf45cab98b5c0b5d00f761 : FAILURE in 9m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc80512f1b6b4cf1b617cdf855065bae : FAILURE in 8m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c71dbaeb4cb431cb87612c067ef806b : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef504309cdac4fafaf8c1227c3854d75 : FAILURE in 55m 59s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f31d615042439d80d063572c74a09d : SUCCESS in 39m 09s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61f709d19c1946fc969cc548fcc49add : FAILURE in 6m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/37222eccd3734cf4858928eb8689e94d : FAILURE in 10m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71e59b73515948baa07d2bae82a970c7 : FAILURE in 10m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be03daa4e57f460da4e6c32b0dc29761 : SUCCESS in 1h 05m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fbdbc9f42d043a992934e690d433fb9 : SUCCESS in 41m 07s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/661cdd1f1dab42feb3be3dd3ae94e455 : FAILURE in 11m 11s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/c93a2fc53506454a898fca68a25f8aa3 : FAILURE in 7m 30s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d9a9188ecdb465198a69de468d61030 : FAILURE in 5m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7165771426ad424ba4693a9f2800d84a : FAILURE in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"19fa07498dce3a9d03f33a13b9af7d360b8cb1e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 09:30:06.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"11daf6982c3ee09533a3f26a86abb63b4606e9a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 09:51:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d0fff4762e05e5edfd101d84175901ddc686fa68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 11:02:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9e860c19289445695dd0bb1d9ce1948\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55d6c426d57249559478615431d8e668 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8162eb143c9541e4aa752778bbdb6900 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/57dcc25734974a12a7fde23a3a31ea7c : SUCCESS in 6m 13s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/120353b270d5427dad2cf0328684ff84 : FAILURE in 12m 12s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da81d27ae71349c1b5c7bc79178de1f0 : FAILURE in 6m 37s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa85bb35c080499a83884841cf7d81a2 : FAILURE in 12m 14s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42b75c093a4b43c6b32385379a937205 : FAILURE in 6m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffb99103cb5d48c198e0bd8f4c051ce1 : FAILURE in 6m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9568f197b7e440df910d38d509539c88 : FAILURE in 7m 18s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4756959f2804802853b7dbb93752e8c : SUCCESS in 49m 52s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a515f16c218f42e48dc5e3da442f790a : SUCCESS in 40m 42s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/615b3ed1d9fd4d13a735a8b2341a0e2b : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0069bd2eea23475c9229dd739626aefb : FAILURE in 12m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac03e0f280da4efba30067df96d99a05 : FAILURE in 11m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/597fa9c7d2a8497a97111be6963b240e : SUCCESS in 1h 06m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d711a0321f814a5e82dc39ff0735f9f7 : SUCCESS in 58m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b87acb122001423b930d552be0a5f7bd : FAILURE in 9m 19s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/d63a6ccab238447280177c3dc53cffb7 : FAILURE in 7m 43s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8da48a79e72742cfa2593bdfc98f8dae : FAILURE in 8m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6cc81a20d9e436da435f51e672efc8b : FAILURE in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"24ea0e2a295be146a2110eee64e556934693a6c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 12:54:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2c66c673b7ee6a28f57c4826bb92ccf00a684590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 14:10:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48576a56464f4db08d7454dfd20a7e7a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b37ce5703014af8accbd63198c7a3bf : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0dedb61f1e54365bc075e7652d8f44b : SUCCESS in 2m 39s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7de0073566a34f83a47698b8583d5b9f : SUCCESS in 5m 46s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/406e4ee0902849c3a00b52a6c97b1c48 : SUCCESS in 12m 13s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/388022bf7ae14866b0054be85956cd53 : SUCCESS in 8m 55s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d5b15d45c1d45c48a5d32a3b795f21e : SUCCESS in 9m 25s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fa8566c14924ee3867c29056d652609 : SUCCESS in 9m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3dbe73f7744abeb19c59b27ed2813e : FAILURE in 7m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fafc620383746f39ef33d2148acfb41 : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f40b91b6674413b96c887c3e5816c14 : SUCCESS in 53m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbff69812e324f2da490085bd491e9d1 : SUCCESS in 43m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ba61c048e014462ad4bf30ae87da10c : FAILURE in 6m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dc7b3f3c5f14acf851114d63d9396ad : FAILURE in 11m 51s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a3fc849c7e045c49cfd6b72426e7489 : FAILURE in 11m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ecfdcaa68d8445bbffdb00c5186b97b : SUCCESS in 1h 08m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/454643c89f2c4c9682bfa59c42dbe4f3 : SUCCESS in 55m 02s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3d421fc5500431986a76d650d001c27 : FAILURE in 11m 41s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb26585a808c48dea38d018ee4fef972 : FAILURE in 9m 04s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/399728df06244521914a0328a84fc572 : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4881531b03d4134947ac2e1ef12d5ed : FAILURE in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c4ceb5c73deaa6814d002a1a80222e5e0003b47a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 14:28:48.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"5590c0d0f6c36d6ce0d508b8cf613494ea0c2afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 15:31:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/596839a8563344699755b6e72213b04e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1942a51f23604f57936e6b16f0360079 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0adf154d4a34b5e92296c44719a98ca : SUCCESS in 2m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdf7971b61ec45ca845892f713c36dad : SUCCESS in 5m 43s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f4132557fad4341baaf34f89ae3cfbb : FAILURE in 12m 08s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0486d1ab0d6a48aea320220153184466 : FAILURE in 8m 41s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7312a04031147e39d1aa382f9a69a66 : FAILURE in 12m 40s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f456185d33f342dbaf21725cf26ceabb : FAILURE in 9m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e1bdad09e4744f78620a728133b2d3b : FAILURE in 9m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09b943611b9243c09a1dcc4c92115b5f : FAILURE in 9m 16s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e6e3f982f394614b1f0d2ef77f9009f : SUCCESS in 46m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31b0a75848cb4df7a6a5d5841aa51b6d : SUCCESS in 42m 15s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5815d1dfc5f5406db6f930bca432c94b : FAILURE in 5m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c56bea59ad4b437f8bd2a7f10a2b6528 : FAILURE in 8m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9eb37f71399d49bdb9eeab813d21de23 : FAILURE in 12m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e28d7e4b92534893aa8a9a3593af1166 : SUCCESS in 45m 40s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35f4851bac8d4046b21c65167e1c20c7 : SUCCESS in 57m 05s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c85cde9014a485b8be0efedb5c6091c : FAILURE in 8m 31s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7faa87f16e034340b7bea7039da58d88 : FAILURE in 7m 23s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc2f1028596a4bebb87073ffc9e98738 : FAILURE in 5m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ec9536fe4a54c53a2400eddc159feab : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"748f476d9a9f9d6e56e44e57c908c55bb23ff5b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-26 00:43:29.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"d8438b6678ecabc9bb28d2ea9bf918e212f5db93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-26 01:57:48.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fc46181aae84113b0da1dbef94a0fdb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da6d8dd251b14d59b7bf9792f6b64394 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7aa537daaa4d4f96afd8c234e336d728 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2489eee47a214158b02d1952c6ff1897 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/63c2a60761d340b1be0e4307bf199fe6 : FAILURE in 12m 16s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e86eab0b5c142c5b5f61fa8bef2e441 : FAILURE in 10m 34s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/80937fc391b94dd58a003e5cd5d4e115 : FAILURE in 12m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1c255ae01e240ed91322a63e4ed227a : FAILURE in 10m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a721f91aaf7144cbb4cb662504d094f6 : FAILURE in 5m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97b6ae407dbe49508d5e97a8eda3d988 : FAILURE in 5m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f004e948d2942deaa51ec5df0188e93 : SUCCESS in 40m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdf47aeb6a674468967771ae3ef6284c : SUCCESS in 32m 58s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dd27c92825a4e1b984b10b1e4a93e18 : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20c720e8957c4fb38a0eedca5f6f8d73 : FAILURE in 11m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3203f61eedd41379dd331e61729ae50 : FAILURE in 11m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d571c6f587df4f11bc6076cde9238320 : SUCCESS in 1h 07m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dafecc9a7954466b4d8a9a3ced58e4c : SUCCESS in 57m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5f46f7625ef46a7921dedc755015299 : FAILURE in 12m 17s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e91ad64398114c03ab027d675cec0fd0 : FAILURE in 7m 41s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d676200fdec4a689b078f564c3c5642 : FAILURE in 6m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d11ee30301f4093802cc2e32d528b58 : FAILURE in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"83fdc7c8d53b1a6fbe98023a416695be3ddce31d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-26 03:07:37.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"c3366353994431cc73f61999b6078debdc321f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-26 04:12:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa9bc0b3b9274affa5dfb41a022439e8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/178a3ed83b664af591809d1fdec07112 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84051168f565490199d24dff2b1cfa55 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/def753c483aa485bb8c4a98eb91bf6fc : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/33347cd999a54816b36bc964d9004944 : FAILURE in 14m 02s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1775257dc6014c13b5ae7a3297aba965 : FAILURE in 8m 28s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2a4c7ec864f45978192632d35e37002 : FAILURE in 13m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c33f6856062c4630a9211563960b05be : FAILURE in 9m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cdd6b368583470b97d4d3f127b33049 : FAILURE in 7m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a5321f6c6940b6b9feb56b12e47af8 : FAILURE in 7m 16s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4149714239a40dbbdecdeb8dff90c40 : SUCCESS in 54m 37s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ef05c15b66742fb90660ba2f55b6c85 : SUCCESS in 46m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/544efe0e61c6407c864ad209d7a556c4 : FAILURE in 6m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ff55110016e4b07b1386f99ded26820 : FAILURE in 13m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/41c3db065e4741e6b84446f0543577d5 : FAILURE in 11m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/76a4003ced5d4089815497c1f68c7051 : FAILURE in 48m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/62ed5a953ac74345af583bc9ad5f1ea2 : FAILURE in 41m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/41a0201ac6a44c78bc9298ae4c30bbdd : FAILURE in 47m 45s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/78d8616e5f6041b39cdf46d52fa01c06 : FAILURE in 9m 31s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a0e6019c1647bbb28a70a309f7d79a : FAILURE in 9m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/beb8bec4ea3a4328bc740966066ff63b : FAILURE in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a28ac61cfa8b7dda9ef20ee028cda671fc36857a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-26 08:24:24.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"4b9519d407aeb6330c941f875a7d232e569cb3cf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-26 09:01:07.000000000","message":"Topic bug/2047326 removed","accounts_in_message":[],"_revision_number":11},{"id":"8646d998241291e3811764097ee2b0dc0b0c38bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 15:50:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"891872504b34d63936d77f551aa6b7f5a5849808","revisions":{"7039ae70b166f7bb685aca550effb59bd5f8c290":{"kind":"REWORK","_number":1,"created":"2023-12-25 05:47:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/1"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:56.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/7039ae70b166f7bb685aca550effb59bd5f8c290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/7039ae70b166f7bb685aca550effb59bd5f8c290"}]},"branch":"refs/heads/master"},"ac5e19bdfffee01f9cd36f096eb12ff7634ba220":{"kind":"REWORK","_number":2,"created":"2023-12-25 06:46:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/2"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 06:46:45.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ac5e19bdfffee01f9cd36f096eb12ff7634ba220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ac5e19bdfffee01f9cd36f096eb12ff7634ba220"}]},"branch":"refs/heads/master"},"ade6bf1792ac721544a9c932344174932f7626bb":{"kind":"REWORK","_number":3,"created":"2023-12-25 06:56:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/3"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 06:56:26.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ade6bf1792ac721544a9c932344174932f7626bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/ade6bf1792ac721544a9c932344174932f7626bb"}]},"branch":"refs/heads/master"},"69d126fc00ea2ee9f381bf688f4addfed8d2dc67":{"kind":"REWORK","_number":4,"created":"2023-12-25 07:07:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/4"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 07:06:53.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/69d126fc00ea2ee9f381bf688f4addfed8d2dc67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/69d126fc00ea2ee9f381bf688f4addfed8d2dc67"}]},"branch":"refs/heads/master"},"76937008562a914ebc68b5f45f3b0f5095f9908f":{"kind":"REWORK","_number":5,"created":"2023-12-25 07:41:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/5"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 07:37:42.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/76937008562a914ebc68b5f45f3b0f5095f9908f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/76937008562a914ebc68b5f45f3b0f5095f9908f"}]},"branch":"refs/heads/master"},"69ecdf7428e615057c14e2433ebde2b86548a950":{"kind":"REWORK","_number":6,"created":"2023-12-25 09:30:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/6"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 09:29:50.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/69ecdf7428e615057c14e2433ebde2b86548a950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/69ecdf7428e615057c14e2433ebde2b86548a950"}]},"branch":"refs/heads/master"},"e93092c05ed3d49ebb0b7764c411100a026b7d1c":{"kind":"REWORK","_number":7,"created":"2023-12-25 09:51:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/7"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 09:50:31.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/e93092c05ed3d49ebb0b7764c411100a026b7d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/e93092c05ed3d49ebb0b7764c411100a026b7d1c"}]},"branch":"refs/heads/master"},"3335ea1a5693e752602addba13375fdd4eea580c":{"kind":"REWORK","_number":8,"created":"2023-12-25 12:54:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/8"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 12:54:06.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3335ea1a5693e752602addba13375fdd4eea580c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3335ea1a5693e752602addba13375fdd4eea580c"}]},"branch":"refs/heads/master"},"07af82b255367dc1d7067939f60d081a3db21e3b":{"kind":"REWORK","_number":9,"created":"2023-12-25 14:28:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/9"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 14:28:37.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/07af82b255367dc1d7067939f60d081a3db21e3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/07af82b255367dc1d7067939f60d081a3db21e3b"}]},"branch":"refs/heads/master"},"874773764fec90b5902df9e527c672a33918fe35":{"kind":"REWORK","_number":10,"created":"2023-12-26 00:43:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/10"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-26 00:43:18.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/874773764fec90b5902df9e527c672a33918fe35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/874773764fec90b5902df9e527c672a33918fe35"}]},"branch":"refs/heads/master"},"891872504b34d63936d77f551aa6b7f5a5849808":{"kind":"REWORK","_number":11,"created":"2023-12-26 03:07:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/904318/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/904318/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/904318/11"}}},"commit":{"parents":[{"commit":"0f500ef7da98b9579f801132f70e030e8d56922f","subject":"cinder_type: Fix regex for properties to accept short value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/0f500ef7da98b9579f801132f70e030e8d56922f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 05:25:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-26 03:07:23.000000000","tz":540},"subject":"cinder_type: Accept hash for properties","message":"cinder_type: Accept hash for properties\n\n... in addition to array, so that users can use native hash value\nhandling to compose up the value being set.\n\nChange-Id: Ibfa6e1aa5e3ef3e8038da611270ebb7f4862f8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/891872504b34d63936d77f551aa6b7f5a5849808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/891872504b34d63936d77f551aa6b7f5a5849808"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
