)]}'
{"id":"openstack%2Fpuppet-cinder~912083","triplet_id":"openstack%2Fpuppet-cinder~master~I775b4ee578169d3c6dfd2087b5286ff6a84de012","project":"openstack/puppet-cinder","branch":"master","topic":"deprecate-vmware","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-08 15:55:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I775b4ee578169d3c6dfd2087b5286ff6a84de012","subject":"Deprecate support for VMWare storage driver","status":"MERGED","created":"2024-03-08 05:57:41.000000000","updated":"2024-03-08 15:56:42.000000000","submitted":"2024-03-08 15:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912083-deprecate-vmware","meta_rev_id":"eba0976d5cb4a83bb755f023a19e99d868f6cd31","_number":912083,"virtual_id_number":912083,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-08 15:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-03-08 13:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-03-08 14:32:00.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-03-08 14:32:00.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-08 07:16:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-08 13:04:37.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-03-08 14:32:00.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"05211f92bc518aff4bdefc05643e37630212982e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 05:57:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"067f401f3fe4c78aa647b522521ccbbcd239923c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 06:01:35.000000000","message":"Topic set to deprecate-vmware","accounts_in_message":[],"_revision_number":1},{"id":"301ac76c565a04876596c7d14a277bfd2a8e12b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 07:16:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7234443bb245c7b020303e2aa6ba78\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bd1c6d982f84fad8bd8c6dbdaf77284 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce1c86bac2ea46eaa98f7538d3e04c29 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d79dced842304e5e9a62e33cb3b0312f : SUCCESS in 5m 12s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/83144b13f889463297d68dad1d776b72 : SUCCESS in 9m 49s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/553cf731b50f42378e581425c63b0600 : SUCCESS in 6m 29s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d1fecd2d21a4854a78a3a3987bc6c2d : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c42fbc3a81549918fb6dc3a0773053f : SUCCESS in 6m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a855f8aa0d34179872666d49c12b687 : SUCCESS in 42m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99b22d1a060441c09ee2b0c867caf179 : SUCCESS in 54m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/470d0b5be50d482aa391099446b718a0 : SUCCESS in 49m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5c1f87345704d70a4658d2d3b76f02f : SUCCESS in 39m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dbb2ade914744cf8903504d2e42b66d : SUCCESS in 41m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb9dc93904844f36899081e428ec9b4b : SUCCESS in 59m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07d7601f62fc4d03b94de6c2e0198a78 : SUCCESS in 1h 14m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f33b39bc97cb4139b54d133d47df0873 : SUCCESS in 58m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/53c96236bb664b9f85230c5199dcd7d8 : SUCCESS in 1h 03m 38s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11e2c34ce4974d13a1029ecfeeb7f25f : SUCCESS in 57m 49s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6dda407bca47a39fd8b4e14cde04a5 : SUCCESS in 19m 19s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a8e7e4ef518446b849de0814cf229f5 : SUCCESS in 13m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04c31bb12fed4541afb1760593305fbe : SUCCESS in 11m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09dab77ca9bb4d92b7b350bafd8c4aa2 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"55f3008987acd6308e90f8c444eefd5321afa035","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-08 13:04:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a32828f620103822972f3efa9ab515dac0c0b3bb","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-03-08 14:32:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5382770be69f0ac95ff1edb39128fc6ec276d65d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 14:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3242c71d94eafb8a0bbedcc95cb6ae4e685845ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 15:55:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3413c43ecef4bbe9486827c881ee277\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a99a5188e38a4e10b9fd40400f5db489 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa7a03ff5dac432cbdfbcd0aa3113015 : SUCCESS in 3m 55s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/39779ba1e7104e41badd1f9ac4dd05a9 : SUCCESS in 13m 16s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e309b21143424c098c93ac515b2cf25a : SUCCESS in 8m 05s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6aef5f36621d48fa94357823737f09e1 : SUCCESS in 57m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec7b6a27f6ab49b48c7caa9224b6c9d4 : SUCCESS in 1h 17m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad7f559af037499ca7319fcdf26c77f9 : SUCCESS in 56m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b88ababab464bba9f7b409b034cab60 : SUCCESS in 52m 21s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/895ffb3d1ff943dc8c078eb5f14d2449 : SUCCESS in 54m 18s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbb561184a0443459f5f5e7127fe4f97 : SUCCESS in 17m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a5d9fc460f345d89b8bb67b650438c1 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cd4ac55ecdb5dc195850cf6cfcd0a36e08611f91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 15:55:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eba0976d5cb4a83bb755f023a19e99d868f6cd31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 15:56:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5df99dcc17bd4f3387e54ede034bb61d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7af96701efc4647beafc0a95f58a6a5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd66bd64465f36bc87ff17969e47c9a0b538bd16","revisions":{"fd66bd64465f36bc87ff17969e47c9a0b538bd16":{"kind":"REWORK","_number":1,"created":"2024-03-08 05:57:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/912083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/83/912083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/83/912083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/83/912083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/83/912083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/83/912083/1"}}},"commit":{"parents":[{"commit":"24988ff24b73b2f65a62f6bcc1193647a03b3a6f","subject":"Merge \"rbd: Stop overriding CEPH_ARGS environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/24988ff24b73b2f65a62f6bcc1193647a03b3a6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-08 05:50:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-08 05:57:08.000000000","tz":540},"subject":"Deprecate support for VMWare storage driver","message":"Deprecate support for VMWare storage driver\n\nWe deprecated vmwareapi virt driver support of puppet-nova during\n2023.2 cycle[1] and then removed it in 2024.2 cycle[2] based on\nthe driver status.\n\nBecause we don\u0027t expect any users may use VMWare storage driver without\nvmwareapi virt driver, this deprecates support for VMWare storage\ndriver. We aim to remove it in 2024.1 cycle.\n\n[1] 8bf7f7d2e9604e24b6226077fced4af92dacd7a5\n[2] 1d0bee66fce81f1d9e40b66f280584ea28e69b45\n\nChange-Id: I775b4ee578169d3c6dfd2087b5286ff6a84de012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/fd66bd64465f36bc87ff17969e47c9a0b538bd16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/fd66bd64465f36bc87ff17969e47c9a0b538bd16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
