)]}'
{"id":"openstack%2Fpuppet-cinder~912818","triplet_id":"openstack%2Fpuppet-cinder~master~Ifad53c99bf84d7109ca7bf8ad3dc0e331fbb923e","project":"openstack/puppet-cinder","branch":"master","topic":"rabbit_transient_quorum_queue","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-14 15:21:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifad53c99bf84d7109ca7bf8ad3dc0e331fbb923e","subject":"Expose rabbit_transient_quorum_queue","status":"MERGED","created":"2024-03-14 00:19:29.000000000","updated":"2024-03-14 18:32:53.000000000","submitted":"2024-03-14 18:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912818-rabbit_transient_quorum_queue","meta_rev_id":"f63a6d80a70520f55b64545ee33082454fc18664","_number":912818,"virtual_id_number":912818,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 18:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2024-03-14 03:24:48.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2024-03-14 11:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2024-03-14 11:05:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 01:50:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 03:24:48.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-03-14 11:05:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9b23cc4620050a5ee8e015214cd78600d0fce16a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 00:19:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f32818ee25e2e7c5ef5cae634d68fd63d9cfb39f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 00:22:12.000000000","message":"Topic set to rabbit_transient_quorum_queue","accounts_in_message":[],"_revision_number":1},{"id":"3f6ad87b97f2f0537ea57679cc13bd36452f0ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 01:50:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec19168049a541d8aa7c6c69fb63788a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95609e7e18bf4532aa32e5606db15e6a : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a64e060888343cdbbfca52ea1af1578 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b17e2c76f9c8462c8278a555d8cf79e5 : SUCCESS in 6m 26s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b570e4cb27b4493a277829fe8d5e052 : SUCCESS in 12m 39s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2dfbd6c68954367b211fe2df4e4a05d : SUCCESS in 8m 59s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dceab463c1d4a37a3270967f9f67ee2 : SUCCESS in 9m 40s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f695b3e09827442eb18bc4c4501e968c : SUCCESS in 9m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3d3ee81ca6a41abb818a585d8b9f8e9 : SUCCESS in 42m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0c6f55292284e5b959327149aab8d72 : SUCCESS in 1h 00m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51d71dd0302342778c7958a447134814 : SUCCESS in 38m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8116c32b99ac4e88888bac2fd10ebca4 : FAILURE in 47m 50s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b09da3c6c2e42c8bff47562787c443b : SUCCESS in 38m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6809075cce854476af2ff7052db368bb : SUCCESS in 55m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c415374c88c745e58899d743eed74239 : SUCCESS in 1h 10m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6924fc9adee42b8b3bed9fd883f8400 : SUCCESS in 49m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b008890dcb149b89802b31aecfa1266 : SUCCESS in 45m 46s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a32c6c112e7f4022a3ac7591311913f7 : FAILURE in 1h 10m 08s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d38aec5fc00438caab7d01d2556cf94 : SUCCESS in 15m 29s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf65b40b5e0d494ebb9afb646b44180f : SUCCESS in 13m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f929491b54d4cf6920581d02e5a8408 : SUCCESS in 10m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90ce6afc7824484583a4619aa81f6d17 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"92f938aa3e7a55a92ae4d1dce8a6e2c852ed52be","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 02:01:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8978be8b6787557b6211c977bee4b8adb82800be","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-03-14 03:24:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"220f3e485c84a57dfb346fc4c38a42f5a95d56d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 04:06:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/033e5beaeb474b28b26773485074828c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/698acd8a1f0c4450bf839ce362b7d363 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3343689581dc43e18b5d2c21e3b95233 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/320ccea21e56402eb3c7a00fc5e6b038 : SUCCESS in 5m 49s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0713f37bd54c494eb96c92786395fb46 : SUCCESS in 11m 58s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27bfdcc6004c4f64a8911c016c3043b7 : SUCCESS in 8m 48s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/3daf15f1870941e9b58ac5194b0b7919 : SUCCESS in 12m 22s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6476865f2844d7d8423ea3ed20ea0eb : SUCCESS in 9m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f1cf7b7308f4cd09326e2c768042fbf : SUCCESS in 46m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f447c8154f9c46cd8907bb97a0f2c463 : SUCCESS in 58m 37s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4266a409484f4915946529950bf49089 : SUCCESS in 46m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4408660199b4228b9664421b3d4633f : SUCCESS in 42m 26s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/929528f9cc62422ebb9e3ee5682d98e1 : SUCCESS in 45m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3bfca4ee64c45c9b6626b6a0f6cc132 : SUCCESS in 57m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbe2bfd2a5034f81822261c5e9d0a243 : SUCCESS in 1h 17m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c06761d3024943f1900f147b55021fd6 : SUCCESS in 59m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6864b7b904bb491b9a258f1fa345be45 : SUCCESS in 49m 11s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40746273fb824bb485a6b008cea32ced : SUCCESS in 53m 56s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/26e4c9cbf16f4988ac3de9c817093395 : SUCCESS in 17m 19s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8716960014945fb97f6c65975c84916 : SUCCESS in 12m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc54cf722d494723ba2eb9b6caeac973 : SUCCESS in 17m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e8a2b7752bc44ea8ae2a7b81a85c3cc : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f3e80007ccb5e183d2c92273e6e5106a775d2697","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-14 11:05:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfdd4d11d60062e411b07cc3080318d3774cdff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 11:06:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d156dc978041092e7e5c4913cf739e7e604dc558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 12:34:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12d2b74c55104921bfd00c32a7835ffd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47321b18cebc40208b457d6a7e6f3f7f : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1e14206080f493888432847186e09be : SUCCESS in 3m 20s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/03f14da5bc02463591f2e8fe670443df : SUCCESS in 10m 51s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56f863506515452c95fb503ec970a351 : SUCCESS in 8m 08s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/21684053f14046dbb8c97829b5cf95a5 : SUCCESS in 1h 01m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68242ffa5fcf4c71a5c788447fd20355 : FAILURE in 1h 23m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0295bc3969b4f628aa3f0efb9db206a : SUCCESS in 53m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9ae525e20a94d9e9aa6d88059a5806e : SUCCESS in 56m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61e74149311c4352bbf6ed28de9fb4c9 : SUCCESS in 59m 00s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/3199a28a5b90445cbf46612b4f606822 : SUCCESS in 18m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9caaecd14213476aae65b5c5cbe322d0 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1a54498e1fc8a3c1a92ff4069291155c742b9fba","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 15:21:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b432594e5ac03cdd319be41a8b73773586ca5f49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:56:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4e41fa3261d4ee5956f2128dfa3b5ff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6844a51cd94f402aa190dcedeb52503b : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89031685ab8d46469f72050b25f2b9f6 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8561a17e582e4b9da4b0d6917b2ed778 : SUCCESS in 5m 56s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/81b688af14ea40f388cd5c6625a916c2 : SUCCESS in 8m 56s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b392a7e007bc47ad8eed4c04046bdb8d : SUCCESS in 8m 16s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb1a16bc25d54b18b15201010588835f : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6d9c20885654b38b03f54f3a44823e4 : SUCCESS in 9m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78bd2721c1454c40b77a52e64532eaed : SUCCESS in 44m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/425408cbc5724cfb8fc4382efdfb7cae : SUCCESS in 1h 06m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81892cc3263744f7b01f952b3368c6a7 : SUCCESS in 57m 04s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84d246b508cb4fb5879e5533b27bba67 : SUCCESS in 39m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e39d095a7ba3448693647cc750ad4faf : SUCCESS in 50m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58dbf1e44ded422f983884f74abaf89f : SUCCESS in 46m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0edd5b31bfa4405cbf7b5aee06c9bc5f : SUCCESS in 1h 27m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8221e1c401de45a1bff849e8bd5ffcb2 : SUCCESS in 1h 08m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68b4491e036c4d11a260dd985eb0cf0a : SUCCESS in 1h 04m 01s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/247e65fd5e624ff2b79b6a0dc35c794a : SUCCESS in 1h 05m 20s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a27b6058ec4e4b64a87305653cbc622d : SUCCESS in 21m 17s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd75f2ed41bd4a0caa6e16f48decb8f8 : SUCCESS in 12m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eae5472a9c944afb9c4cb9d3a60bee35 : SUCCESS in 12m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cf44ad4256244ecbac014eb1537883f : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9bff16323a3e3a760d70350a0d2e8aaeee6f737c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:57:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40c7c2fa6acebe2ed2d2a2fcdc8eb884e2fab5b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 18:31:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4a5c7a0a4b94b29a6d1043a44586788\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f54ec9418c024aa8a6b40a187c44814f : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/249dced95cb541b39fc8e6d0be121df3 : SUCCESS in 2m 35s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/5314b77d1648418897fc6c5ddc52475c : SUCCESS in 11m 58s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/927dbe006f6e4ec7bfb23dfe27964f90 : SUCCESS in 8m 39s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d31d569bd4542ebac29795754b78499 : SUCCESS in 1h 03m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1f5daa648e94c78b9420344a18d7277 : SUCCESS in 1h 29m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b662d4db7bd4a0a82953b35dee4f78e : SUCCESS in 1h 04m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/449a6dc9fe884e33932a5ba8963f9ca8 : SUCCESS in 55m 41s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1f1d5ffb1ff4f92b6819f007b9ba6e6 : SUCCESS in 1h 01m 58s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/9376e966e7624bd7be03030d2037d8f8 : SUCCESS in 20m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/740f14c5a58543c9a69090b1032cbbcc : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dbbf75c4a94bdb8bb0fbe081c4344c9cc80feebc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 18:31:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f63a6d80a70520f55b64545ee33082454fc18664","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 18:32:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65eb55f7faa4563b5b0cc8e4a4de6e3\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4696c338aa32450ca9381b31ce0206bb : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23634ccf2579def1cb28d87155fd6a203cc2c2f0","revisions":{"23634ccf2579def1cb28d87155fd6a203cc2c2f0":{"kind":"REWORK","_number":1,"created":"2024-03-14 00:19:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/912818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/18/912818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/912818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/912818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/18/912818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/18/912818/1"}}},"commit":{"parents":[{"commit":"24988ff24b73b2f65a62f6bcc1193647a03b3a6f","subject":"Merge \"rbd: Stop overriding CEPH_ARGS environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/24988ff24b73b2f65a62f6bcc1193647a03b3a6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-14 00:19:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-14 00:19:14.000000000","tz":540},"subject":"Expose rabbit_transient_quorum_queue","message":"Expose rabbit_transient_quorum_queue\n\nDepends-on: https://review.opendev.org/911021\nChange-Id: Ifad53c99bf84d7109ca7bf8ad3dc0e331fbb923e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/23634ccf2579def1cb28d87155fd6a203cc2c2f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/23634ccf2579def1cb28d87155fd6a203cc2c2f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
