)]}'
{"id":"openstack%2Fpuppet-cinder~914578","triplet_id":"openstack%2Fpuppet-cinder~master~Ibb49c49f85f0c5bee8545e240fe3c6aac7e3d13b","project":"openstack/puppet-cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-01 23:09:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb49c49f85f0c5bee8545e240fe3c6aac7e3d13b","subject":"rbd: Simplify handling of rbd_cluster_name","status":"MERGED","created":"2024-03-28 06:17:47.000000000","updated":"2024-04-01 23:09:18.000000000","submitted":"2024-04-01 23:09:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914578","meta_rev_id":"fdf9093ef4137fb54bea27d41600c9bfbf7b7ace","_number":914578,"virtual_id_number":914578,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-01 23:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-03-29 15:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-04-01 21:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-04-01 21:39:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 07:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-29 15:15:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-04-01 21:39:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"77ee8396144b58f9e32ac80e9a5fc751ae5a9bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-28 06:17:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e99d0af807b698ed62084af6c8495a979a70e9b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 07:13:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e66f94aeebe48608526f6074b6677e1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca0170cee5c94317bfea8cf31faa4775 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/780afca553114a9896a1a9e8bb9d2faf : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/5febaf232c964c2b9406b41190551500 : RETRY_LIMIT in 2m 28s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/367a9e3f6d1e429fac217139855a1839 : RETRY_LIMIT in 4m 24s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19c4e0c2341b4d80a889df084de3a7d6 : SUCCESS in 8m 41s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/329f97a1275c43b09ae17f798c97d00e : RETRY_LIMIT in 4m 30s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5200501f70df4aa6bad3ad1424151e5c : SUCCESS in 6m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1df76eeeda304ce6bc46b6e6ac0c049f : SUCCESS in 43m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a833822231c341b280f1f9924103edad : SUCCESS in 53m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/363e9d8ba2d24012bd20360fe0e6ef45 : SUCCESS in 44m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee22937ff7694d69b032212500cba168 : SUCCESS in 40m 48s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07d48fe09aa644e7954a692e69657119 : SUCCESS in 41m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/87a32672fee7412f87879544b8292f37 : RETRY_LIMIT in 6m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/958441638c2042d098056889092d9f45 : RETRY_LIMIT in 4m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4465315a683549e99667f8452cf07d0e : RETRY_LIMIT in 4m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5368b5196c9b4ee98cc2f463acc46256 : RETRY_LIMIT in 4m 44s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/308dfc8b1cfd484aa4e4d1a460420a91 : RETRY_LIMIT in 4m 51s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbc69981fe1c4ada9ece7c2f71a36f5c : RETRY_LIMIT in 4m 32s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbad46a60c9e46d6bd4688f12754b084 : SUCCESS in 13m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce533a7c71574b8ab34c3ffb877f8a46 : SUCCESS in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"595a2fd332633a17683b3fb8df692f35e53587c1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-28 08:17:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69ce714524f3e1bb1267d8ffa67956fcf42e476e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 09:25:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/352d8e3d25ba4f12bb18b928b9b00e11\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34e1d64ecaae4275acca74a34272a2c8 : SUCCESS in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bcb66f5cc1f47419d5c0a19490ba0ba : SUCCESS in 2m 22s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0179c02531454b369fb306e582e4746c : RETRY_LIMIT in 2m 36s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6a7a37e82b24fa3bb4b1341c0bbee5b : RETRY_LIMIT in 6m 49s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4e24e6dee95444b9b1cbfd0555921bf : SUCCESS in 7m 56s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/f92541a9f8b946c2ad2893c9c317a1e0 : RETRY_LIMIT in 6m 47s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd0978715ee8408a99dd9ca41ca35995 : SUCCESS in 7m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/290a4415f6e24ffc9f274caed516d743 : SUCCESS in 42m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3df037eff674a61979b0a9376ced0b0 : SUCCESS in 50m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d60224efbb9348e598b3f8c8a71150c2 : SUCCESS in 47m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cfe96e0e09b47a080e773471d917e86 : SUCCESS in 37m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d20c1472d86a4537ab5afca65769b07b : SUCCESS in 43m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d4d1db001554f218cbbcc3e17eb9ebc : RETRY_LIMIT in 5m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0608a36a9afd456b85e123541d481edd : RETRY_LIMIT in 8m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7e42a7cfde64d23bde1942e719abd76 : RETRY_LIMIT in 7m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74716ae5dc574981afd4ddfbc22b1899 : RETRY_LIMIT in 5m 04s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dfa35e87b4e43f4965d493f8eb3497a : RETRY_LIMIT in 5m 02s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/e398eb4729d842fd881bb1555eb8ef59 : RETRY_LIMIT in 4m 44s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f57ff387098c4a6bbd572b13a24000b4 : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3469522796c14cdea1122175b6474ded : SUCCESS in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd8ef574502a9cd77b222a7ec99ae996373aa96a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-28 14:24:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae0d6b0388f4a6ce667d369cb4850580af692a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 15:44:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c899cd9ea43040f1867048a739a0770e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8561bea53f4841269f526408167332cd : SUCCESS in 4m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33d6512a7f5843aa9c2a50dcd33cc00c : SUCCESS in 3m 08s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4405428cb12f4095aa4891e6d56222ea : SUCCESS in 5m 37s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ef2a34da7da44809b826f0de281c007 : SUCCESS in 12m 01s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df0cdd931e88429d9c5c929b780e2303 : SUCCESS in 9m 11s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/38e2a319d4504fbd9ee9efa3324bb8f8 : SUCCESS in 12m 41s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65ff1141148e43f9810b188c5106bca7 : SUCCESS in 9m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92b1dc35563f494c833e2bb461648973 : SUCCESS in 45m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df55a98308c9412ab5789e680b5b3815 : FAILURE in 59m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ca812664a54bf28637450efca2b0be : SUCCESS in 48m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5508a90b0fc446be833b1128d1b2f331 : SUCCESS in 44m 05s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8ccf2548b814f7e9c138fba2cadb2bb : SUCCESS in 48m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce4e158f44ef4218a689273844a7e246 : SUCCESS in 1h 00m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1d64b4987be4589801f286b7097fbdf : FAILURE in 1h 12m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c39192f8f84c1aad5b0d6c3b8764d5 : SUCCESS in 1h 01m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f29ad78c62c54c3aa0b9245048fa2ec9 : SUCCESS in 1h 05m 02s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb60eced25444d17939b77123d70d4a1 : SUCCESS in 59m 32s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0a7e2d2effe4192be14b291de63ad3b : SUCCESS in 20m 27s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7386722ef914e729a387f48974aa275 : SUCCESS in 16m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14a934c3ffea4b829189568ce7b9fdaf : SUCCESS in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56af2bc05ae06cb1c483216c5dc4a425ee57cd5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-28 16:00:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9af9541ed0563d0d48755a1097a73021559ddf0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 17:53:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6417bf52635442f58377a6ca8fca84fa\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/622b5a143ebd41f9abd9c934a6d88213 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e618e752c17240a5bd925e9c50a87f51 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d7b511610f74d15a42362a15b594534 : SUCCESS in 5m 07s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3c4ad7315de48fab3f0c04bb47b4cf6 : SUCCESS in 13m 06s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e81ee9c585a4916ac9fe0e661bf9075 : SUCCESS in 9m 29s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/c70c4f3ce8b74bbc8617a21c374aef52 : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fa4b80a29a24f7ea56b5bdff2ac8a86 : SUCCESS in 9m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aac93f3a4c3f45fd9095f0c1704105cc : SUCCESS in 45m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7779ce408ebe49ed83dcbc0a8347e814 : FAILURE in 55m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73c2f389c9864d8290862507db0bd689 : SUCCESS in 48m 56s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cf51a883649417698e1b20567e8bdf6 : SUCCESS in 41m 47s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02a113aea7ea44a38e88eefbc0d760b7 : SUCCESS in 46m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2d1c7d390d34880a496e664b149796f : SUCCESS in 56m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e422d24bcfeb4f579ad5a77018d1c4eb : SUCCESS in 1h 35m 17s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/897245ae7b2b4051ac45cb890805d9e4 : SUCCESS in 1h 01m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/31258a9162314560af777e82552a0b36 : SUCCESS in 57m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4616c19a88c342acaeafdbbf9aa3ca6f : SUCCESS in 1h 00m 00s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/77e7527219ff43acbd3a7b4698f18c66 : SUCCESS in 17m 27s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dea2454a4694b6bab6cc863a5f434f9 : SUCCESS in 15m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecb8b191c85c4db1862d1a9e5051db47 : SUCCESS in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22694997e84c7c9138a1f7a9d82b2b85d4352c7e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-03-29 15:15:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b57399e07234a7552301793b42de095da1ab4ab","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-01 21:39:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9fd4906142c599c75c96a4318755075d461c77af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:40:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a54515400197c164d449eef84f053f295544d9bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 23:09:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81a51c28e06440808712326eea05c7dc\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e03cfa017fea472597f2ec2031e4ed35 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6cb76f70b994dc38d259d80b8ee033f : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/eca579b9083c4b5da7bb8df8b4b9a0bb : SUCCESS in 12m 27s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8239892b252a4530996a42ebdbd44e02 : SUCCESS in 8m 59s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5ba0da65c12435aad06b589874be9dd : SUCCESS in 1h 10m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f426a00ce5647a29828a2ec03791e58 : SUCCESS in 1h 18m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cd1cf7aa2d14732b662e3349301a362 : SUCCESS in 46m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c4580eff75d49e189f63386b31a1863 : SUCCESS in 58m 18s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72b26bf04df8439bb8f4578598eda0b2 : SUCCESS in 57m 09s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebe8b388bb6e402b945f4a46a18e8e83 : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fdf9093ef4137fb54bea27d41600c9bfbf7b7ace","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 23:09:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdc045dba61a2cc9b003b0ea5bf2ca1e72a04667","revisions":{"bdc045dba61a2cc9b003b0ea5bf2ca1e72a04667":{"kind":"REWORK","_number":1,"created":"2024-03-28 06:17:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/914578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/78/914578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/78/914578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/78/914578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/78/914578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/78/914578/1"}}},"commit":{"parents":[{"commit":"a17d67766d2b27edff66b43bfe0ba03d1359eb19","subject":"rbd: Purge rbd_cluster_name by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/a17d67766d2b27edff66b43bfe0ba03d1359eb19"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-28 06:17:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-28 06:17:07.000000000","tz":540},"subject":"rbd: Simplify handling of rbd_cluster_name","message":"rbd: Simplify handling of rbd_cluster_name\n\nWe can accept config change unless we backport the change, because\nservices need to be restarted anyway during upgrade.\n\nChange-Id: Ibb49c49f85f0c5bee8545e240fe3c6aac7e3d13b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/bdc045dba61a2cc9b003b0ea5bf2ca1e72a04667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/bdc045dba61a2cc9b003b0ea5bf2ca1e72a04667"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
