)]}'
{"id":"openstack%2Fpuppet-cinder~929724","triplet_id":"openstack%2Fpuppet-cinder~master~I411e0741a923d65aaca91d5167762f46a4f39e02","project":"openstack/puppet-cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-19 23:54:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2024-09-20 03:27:33.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-09-18 20:51:12.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I411e0741a923d65aaca91d5167762f46a4f39e02","subject":"Expose configure_service","status":"MERGED","created":"2024-09-18 05:56:44.000000000","updated":"2024-09-20 03:28:41.000000000","submitted":"2024-09-20 03:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929724","meta_rev_id":"c6f9d601c78decdf670b362260d230e2fa7206fb","_number":929724,"virtual_id_number":929724,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 03:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-09-18 20:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2024-09-19 15:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2024-09-19 15:40:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-18 08:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-18 08:21:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-09-18 12:56:22.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-09-19 15:40:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"70605df76feecd0579ae1090a746a13c43b2f61d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-18 05:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31774653d3b5684bacc2348c05d6137a8f96a898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 08:08:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f5d99a4c0b94d028f38003ce64108d4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c65581dd0fdc40a592883d4e1279c2b7 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21c78a7ac83147cfb740fbbb3ade0676 : SUCCESS in 1m 56s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec7f3cbd187c4765aa023918937f5ec3 : SUCCESS in 5m 08s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/c92e446f2c1c447b9ff04af4d4d57065 : SUCCESS in 9m 09s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29aeb157161d44c28943f2f158bd0831 : SUCCESS in 5m 34s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c19e22bf864747709fe351e3bdca5265 : SUCCESS in 9m 56s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a18e20da0eeb4a3992fc533bb4796ea9 : SUCCESS in 4m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e26d96b055ab4ac0a3c78a4c7607c250 : SUCCESS in 46m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e9eed7108e14bc887643fc6f8951972 : SUCCESS in 51m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c22c37fc5dd54466a865a1afa450e5e5 : TIMED_OUT in 2h 01m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faeb2aaa00d741ae94c38beb77e77fe3 : SUCCESS in 43m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/098b2b086fd64d4ca70fa3535b4a4af0 : TIMED_OUT in 2h 01m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f70a89d628e4616b873831ba83d1c41 : SUCCESS in 1h 01m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48bd91ee1ea84be7b15587e6b1d25707 : SUCCESS in 1h 14m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43845f7c4d2c4cbe9e7eb5ffa30e1ac0 : SUCCESS in 50m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/946564c2ca844ebca24c6cb31d077056 : SUCCESS in 50m 15s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa41769abce248c2a63b17c04cb7d1e0 : SUCCESS in 39m 15s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ceed4aa05f0e4d3e9a4d88c7704e113d : SUCCESS in 14m 41s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/159d61ef244d4021906c58008e1df35b : SUCCESS in 11m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ebfec63baed47a1a435d1f03b856b0a : SUCCESS in 9m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a37e2c58df2a4963bca7ca3ada2180e6 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ace989bf7ab4f3da0f834035f2d425abbb24819b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-18 08:21:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ed9651622caf593c849e6709141dd8e34298e65","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-09-18 12:56:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"15c11ba751042a2fdb314950d87b196278de0114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-18 13:32:44.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7d0ff3aa199606042d79cc3e9e70a724fe78c1a9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-18 13:33:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28b17ebf34b0b1ffa5f95cddd08c12802f893f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:40:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9978a6af1794edb959abd8dafcd3278\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e7211248391435aad494e71ba8efcbd : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b32e11753d24300abe64cdac751c34c : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c714c8ef5ce841258d51513f2ce958d6 : SUCCESS in 5m 14s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5c95c62b6c54839a159460c22fe2e38 : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc82ea56faf4f60bde933d82df2f308 : SUCCESS in 8m 48s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/628a30197676474fac5a0b7eb75fc6fe : SUCCESS in 10m 43s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6bd759ad0a24ba68f2052959e14a991 : SUCCESS in 7m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5020ba742a7484c992c5ddb3ac49485 : SUCCESS in 44m 56s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11cb2aef3dbf4c0fb7a8d685acc456a3 : SUCCESS in 1h 15m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf6bf85f978443d69187e3af3ecaa39d : TIMED_OUT in 2h 02m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdcc9bdcec5d43539c542f3888523658 : SUCCESS in 43m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24fb65706b3144c49dedd26f5566f7eb : TIMED_OUT in 2h 01m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/55a8571e5f1041809d14416b32bf00eb : SUCCESS in 51m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/10608d7e1efa4725a4b402313081269d : SUCCESS in 1h 17m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f41131fe1a154fc79ecc5db2b1aa8e25 : SUCCESS in 33m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6023e93c54d644eb82714e83adfaa0b7 : SUCCESS in 52m 45s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/313f35e4b17a496b8f6bdc27b15478a4 : SUCCESS in 53m 02s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bc991849143447d803153d220a09399 : SUCCESS in 16m 14s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91cecbd4c46344a4b3a29065cecd961c : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/555dd0254c6a4959a0ee9623b5a8c987 : SUCCESS in 11m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01020e5a31d84215a740ddbf1784e1ce : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"043d8a910593863819044dc529da86f70e882336","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-18 20:51:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4e357873868c1498df74116560bcc1c4a2aa55e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-09-19 15:40:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c2c3babbae95e401ef48b92212169a3257e1acbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 15:41:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2584090c95a351795b78e83d8d54c166df965b89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 17:02:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f0774636fc943cea54a4cfb70a6771c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d5693519e2143ae83d90e7731cff95f : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ba62e1967af47b294bc03aae17f9b65 : SUCCESS in 3m 28s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f20587160a244cdb92bc4fc7526e9495 : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef22b76e98c4429aaf6dbfbffaa5cd06 : SUCCESS in 5m 13s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a0c40ab475a46dca0e8f87f1d9a7042 : FAILURE in 55m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7eecfc9b0db4ed69b33a24c9a3cf845 : FAILURE in 1h 19m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a1927baf57f4da58e4c9b55f9fdd7eb : SUCCESS in 1h 02m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b515d49bb7614015b50e87a7602996f9 : SUCCESS in 55m 13s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/657a11266d5c48ee8a01893d30614ca0 : SUCCESS in 59m 08s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce9d9548a4594bfd8bad3d3b9eb1a712 : SUCCESS in 23m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7498d6e1e1342fbbc0c885fd89384ff : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"24e06606fc60885f3455b9f9222f13cc4e807528","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-19 23:54:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc46f017d0222c1775f6d6e9aaba4e348d8b07f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 02:01:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d115a67f044a318561c320b8b2b172\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e4cffd0faa84959912df22bf34edd88 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/973c964f366c4566becc46bd891e9267 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7439bcfbb0984ecbbecf05a385d41433 : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bb5193586d54e738ac3647efb052df4 : SUCCESS in 10m 16s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5a658e5c0fa458885bd566a7231a9d6 : SUCCESS in 7m 05s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/923c173bffe44986a49b95812726d07d : SUCCESS in 10m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cee3550493440d5ae49dec64df644e8 : SUCCESS in 6m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/460723c7c46a43dd912ca181011efd27 : SUCCESS in 39m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3731fc930d064cb8a87e5fabce8d7b7f : SUCCESS in 57m 38s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71d07cd413394925b15299c6c3824b2e : TIMED_OUT in 2h 01m 10s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af35b947220748b990652c4bdcaee3c5 : SUCCESS in 43m 48s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff7d450fb11a4bda8c336d39f8cc0b69 : TIMED_OUT in 2h 01m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ba984bf9e4047bf84e80285f41dd322 : SUCCESS in 33m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3696396ca5c4a5e9e7e51ff052d7aa1 : SUCCESS in 1h 28m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a0af2289285465284d6da0f687b60ee : SUCCESS in 51m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a207356b3c54d50b646507e9653cc31 : SUCCESS in 46m 39s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a48e4624bb1f480f9e1a8c2f81c79ca2 : SUCCESS in 53m 57s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4112b36c5bd4220a1434fead9a18e05 : SUCCESS in 17m 09s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a7df4b9000a4a91b271f50f7dc50419 : SUCCESS in 14m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f504677946f94bff95d4f8d320156236 : SUCCESS in 8m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab1db399ebfb4ed6a57ec498919ed309 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"711d7bfa3a3386e9fbe9f622f0ed418bbf769015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 02:13:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"777a225045312db10ccb872b95c0b1c0ac21359a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 03:27:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec50c89b3c0a470d8d09dbfd0667b760\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e04812d7e2e417cae4136cc7c20074d : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0879d116ee744cd2ad319cdabeafb25a : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/89e982c54c204da5a851ad8c7c3ccbe9 : SUCCESS in 10m 30s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69635d59aabd49b4866b08c247423437 : SUCCESS in 7m 13s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea96095584b54b938712d64b2385c48d : SUCCESS in 55m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/98c4f40d657547e48353c7d0230dee93 : SUCCESS in 1h 07m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9d5d52be5d44c44ab7aa079f5cbf6d4 : SUCCESS in 53m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3aa8c5586da74afdb8ceeb804be8ac1f : SUCCESS in 55m 17s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c4ac15104e94594b5350a1b810cd374 : SUCCESS in 55m 15s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/926f160b3f6b418bbcae13660abb4e31 : SUCCESS in 19m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbbacd353eed4ae5add53ad4fdca7129 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"76a7f1ae3936b63b1818cfc1b6722916564a2710","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 03:27:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c6f9d601c78decdf670b362260d230e2fa7206fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 03:28:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6881e01cb1e4abd9b070ae2fb353f5c\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d9f4d40b0bc492db419359068ad1d7d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e0cc4f7d4b39ea37b925ec7f5b4b665c5629b07","revisions":{"d91be41432beaae785acdc64130972bd03deb33b":{"kind":"REWORK","_number":1,"created":"2024-09-18 05:56:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/929724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/24/929724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/1"}}},"commit":{"parents":[{"commit":"b897bfdcdb7341ae3073feefcb88da0765025808","subject":"Deprecate management of cinder v3 user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b897bfdcdb7341ae3073feefcb88da0765025808"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 05:56:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 05:56:29.000000000","tz":540},"subject":"Expose configure_service","message":"Expose configure_service\n\nThis parameter has been implemented in the underlying defined resource\ntype but was not configurable.\n\nChange-Id: I411e0741a923d65aaca91d5167762f46a4f39e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d91be41432beaae785acdc64130972bd03deb33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/d91be41432beaae785acdc64130972bd03deb33b"}]},"branch":"refs/heads/master"},"7e0cc4f7d4b39ea37b925ec7f5b4b665c5629b07":{"kind":"REWORK","_number":2,"created":"2024-09-18 13:32:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/929724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/24/929724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/24/929724/2"}}},"commit":{"parents":[{"commit":"b897bfdcdb7341ae3073feefcb88da0765025808","subject":"Deprecate management of cinder v3 user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b897bfdcdb7341ae3073feefcb88da0765025808"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 05:56:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 13:32:44.000000000","tz":0},"subject":"Expose configure_service","message":"Expose configure_service\n\nThis parameter has been implemented in the underlying defined resource\ntype but was not configurable.\n\nChange-Id: I411e0741a923d65aaca91d5167762f46a4f39e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/7e0cc4f7d4b39ea37b925ec7f5b4b665c5629b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/7e0cc4f7d4b39ea37b925ec7f5b4b665c5629b07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
