)]}'
{"id":"openstack%2Fpuppet-cinder~942758","triplet_id":"openstack%2Fpuppet-cinder~master~I2a604308287b6787e55c3fa3a81a3e36486c3a27","project":"openstack/puppet-cinder","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-26 12:51:11.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2a604308287b6787e55c3fa3a81a3e36486c3a27","subject":"hpe3par: Remove own hard-coded defaults","status":"NEW","created":"2025-02-26 05:30:32.000000000","updated":"2025-02-26 16:54:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac40fab075eff1480ca30688d514bab6aa370ffd","_number":942758,"virtual_id_number":942758,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-26 11:41:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-02-26 12:51:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 07:21:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-26 12:51:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-02-26 16:54:33.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"}],"messages":[{"id":"7086d4cefa8d2b31611d3552e965b0173621fe93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-26 05:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9daf81191e363e1eeff4e1706c235e261542405d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 07:21:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ea049d95886464191878bfd4bb625e0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02de7e86485944ea851c565de6dd806e : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/315a510db70f42b083b537d205e0befa : SUCCESS in 3m 06s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4df386add2ab4d2ea6d4bf07a020ab57 : SUCCESS in 3m 21s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2fef9c7eb7440328a0e69b46fe65edd : FAILURE in 8m 41s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/857d969b75504d20a6c65ecdef0c1077 : FAILURE in 6m 18s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddc6d4664f0c4e0fac03e1c9dee758b4 : FAILURE in 9m 07s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed239523177c4945bf80fde6892d638d : FAILURE in 6m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f32dca165bf5424382a49f91ee430ef2 : SUCCESS in 43m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cb713b5b1dc4cf48e49e5bb12c5dec6 : SUCCESS in 51m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d3b28de26614d94a3d85c077eb77790 : SUCCESS in 38m 28s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec92de82a080457885fdffba24f3d112 : SUCCESS in 32m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ad3d5d6d5344c039a1e459e853dc023 : FAILURE in 25m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cd11a6bcbc14077af7f7916161176c9 : SUCCESS in 41m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64f28c41b9a943369164d28661c9fa3c : SUCCESS in 1h 45m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f0160a3474c40b9af3d9ec632692385 : SUCCESS in 1h 02m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05e928e3156e41fea85c7113c781521e : SUCCESS in 37m 28s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/45ca1db680204caf92bfc3cc8c838d4f : SUCCESS in 1h 24m 29s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbf2c5ac092c414a8bcabfd98fe65180 : SUCCESS in 41m 51s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a009bed0b56f483e831d8e0f89126e34 : SUCCESS in 1h 39m 41s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bff4e051f496485488f81235d44f35db : SUCCESS in 1h 14m 15s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7147b59e4ef6409898b167e7de89a141 : SUCCESS in 1h 19m 48s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca7c6bf5cfb14ebeb21f72dd530dacd8 : SUCCESS in 40m 37s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6c6ae44e469429dbbb634968a23ce5c : SUCCESS in 35m 51s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb81c52245e64a1e87704da3e5a8a6d8 : SUCCESS in 12m 26s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/15f699f060b24250b9a5d00b3a4bb98c : SUCCESS in 35m 56s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91a985a6ffdf46c88fd7517636793525 : SUCCESS in 7m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b4042c02f404900ae0600603e291dd0 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"62e3893fe7a3289e9d19d5e561015ce04c7ff6ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-26 07:55:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8fae2523088f0de2a6d8ba9c1d2c7d5a0f1dc558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 09:50:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de3ca7348d8a437fb06a812cb422fe18\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65ae65d913294f66b09aabdb25648708 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4bcf4070dba4aa5ad2df82f6d1d80ab : SUCCESS in 2m 31s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1dd0389271245c49b53b18332e12e9c : SUCCESS in 6m 01s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/593ed654b60b4ce78f6127b016c167d0 : SUCCESS in 10m 03s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61ede47a695b447c8b3645d8eec7f43e : SUCCESS in 7m 04s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f0175ffe8e04e38948feb3db0704641 : SUCCESS in 8m 49s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da25260ab1d84a519c3be36cb7c8cafb : SUCCESS in 6m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21c9cf9ec54f43a5a82f88aa05c39d10 : SUCCESS in 48m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8225b28b85440cc984d21c658e11479 : SUCCESS in 1h 00m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45a314f5c7de4d15a5b1dbd90c3e9663 : SUCCESS in 46m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09c3463f108c47d8b83a0c79d5807466 : SUCCESS in 37m 56s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7681f04e1eb145eba69ad4b96da6a9bc : FAILURE in 27m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58aecde3db0f45f98affb06e27e342a9 : FAILURE in 1h 42m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4195c2decdc84121a6cf97691085d3d0 : SUCCESS in 1h 48m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cd3f20059474afdb7ead94f1bc349f8 : SUCCESS in 1h 41m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb8381ff442f4d9fa05299dd8b318094 : SUCCESS in 1h 09m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b2c7f8e6fd148c98c2137febe28d94a : SUCCESS in 1h 11m 15s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb8f94b97d20403186d38a52c9b2c1dd : SUCCESS in 1h 22m 18s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86e1372cec2b4cd1b889ea4b27e6e394 : SUCCESS in 1h 35m 36s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/62bd2cb59b1d4a1fb1089606dbf52bc0 : SUCCESS in 1h 41m 36s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49f4ed84ec8749d3a7d0b6e1b19a3c67 : SUCCESS in 1h 17m 17s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d649766f1c941e4b95d91f47f6bd3ad : SUCCESS in 1h 07m 34s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce2b937dae82449fb6c31a4187528470 : SUCCESS in 45m 03s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f045fd2be3b34c2e8d352b3c9afb7568 : SUCCESS in 8m 36s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9b30feeb93a419eaa060c98f64d3fae : SUCCESS in 34m 15s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b13d851a8d8f4be88c8390827853e912 : SUCCESS in 12m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08f87fb39e4c449c9c48959924b2d64c : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"44f0a8da0aa2690a66d21b7d5fd4cf4cd6a9b56b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-26 10:00:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"560a5d383de8eb6791e16067061a7b5d822a045d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 11:41:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59ef554b20f14c84bb61309247230811\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c570e4984bde4681b91d46cc293b6c63 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51d29cc1ad94d8d8ce10ae1c36d9781 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/55da7a87876d4f9ab63819379023a522 : SUCCESS in 9m 54s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/7416ebfdbf10405faa12af031f26c483 : SUCCESS in 15m 45s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76be54df5d794d378173c71660604dbf : SUCCESS in 7m 03s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/48e84cfc5e7344b3b21e3b13e8bba66e : SUCCESS in 14m 40s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/485fa51a9ed14b8f86dc95f5bd3dd6ab : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98821593f24f4b589e4c862ef23d4760 : SUCCESS in 58m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b2a7ca5489f4cbb8096bfa0ac1b0b04 : SUCCESS in 58m 57s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccc413f6f8fe4e15b3e06bda0cd8c266 : FAILURE in 30m 35s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54777c9447f947d5996d3278547a6e3c : SUCCESS in 38m 01s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f9453564fe8446f827ea3d896152993 : SUCCESS in 38m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/417d0c49fda6484bbfb2b32c65f65fc4 : SUCCESS in 1h 35m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d11a295def24785a9c7e8777e65f548 : SUCCESS in 1h 26m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e559894b3cd40f1a24fa8a9fec869da : SUCCESS in 1h 08m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30fa1609507e4d0a868f3b43313bf944 : SUCCESS in 1h 05m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86645a7b31f44c0db503ed4cda0f21c3 : SUCCESS in 1h 08m 18s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4074a6b9ac64e969929489016309528 : SUCCESS in 1h 25m 12s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9185ed3fa4404e088b0727bdc93d16b0 : SUCCESS in 1h 28m 00s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2bf45ea4808456095f7c14afb7d6c3f : SUCCESS in 1h 24m 05s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/904dffe6cad14618bcfd42fd8e940c44 : SUCCESS in 1h 11m 16s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/937f4e26b2734725b732d9e793e049a3 : SUCCESS in 1h 15m 45s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/32b2d4107bc8407496a719a07033e76f : SUCCESS in 43m 37s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ceb13595d5a464bbbcc6b07ce60abe6 : SUCCESS in 13m 35s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d42c7d100e0b47f0afb4058b2b5d779d : SUCCESS in 41m 13s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9892dc6973da4ce0a0357e3758697211 : SUCCESS in 16m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55d3b5ec43ea480cae4704916d4f4ec9 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"a32b6cbc74cbf9bd4e5f17d00e2c259761f1869e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-26 12:51:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ac40fab075eff1480ca30688d514bab6aa370ffd","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-02-26 16:54:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3777f97c1325bac0f8e76f25dbaef29ecf0887a8","revisions":{"25f45a3f8c28ff5522342631e8f1f3c6a0446d57":{"kind":"REWORK","_number":1,"created":"2025-02-26 05:30:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/942758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/58/942758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/1"}}},"commit":{"parents":[{"commit":"41facf68a7ca56ec5c464e13dc23a251d965d778","subject":"hpe3par: Replace redundant hard-coded default values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/41facf68a7ca56ec5c464e13dc23a251d965d778"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-26 05:29:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-26 05:30:23.000000000","tz":540},"subject":"hpe3par: Remove own hard-coded defaults","message":"hpe3par: Remove own hard-coded defaults\n\nAs a deployment tooling we do not have any explicit reason to override\nthese. If the current values are recommended by the storage vendor then\nthe values should be baked in cinder, not in this module.\n\nChange-Id: I2a604308287b6787e55c3fa3a81a3e36486c3a27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/25f45a3f8c28ff5522342631e8f1f3c6a0446d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/25f45a3f8c28ff5522342631e8f1f3c6a0446d57"}]},"branch":"refs/heads/master"},"3777f97c1325bac0f8e76f25dbaef29ecf0887a8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-26 07:55:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/942758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/58/942758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/58/942758/2"}}},"commit":{"parents":[{"commit":"6e182455e79dd53fde1955e58a7de6a3e1a6631a","subject":"hpe3par: Replace redundant hard-coded default values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/6e182455e79dd53fde1955e58a7de6a3e1a6631a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-26 05:29:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-26 07:55:35.000000000","tz":540},"subject":"hpe3par: Remove own hard-coded defaults","message":"hpe3par: Remove own hard-coded defaults\n\nAs a deployment tooling we do not have any explicit reason to override\nthese. If the current values are recommended by the storage vendor then\nthe values should be baked in cinder, not in this module.\n\nChange-Id: I2a604308287b6787e55c3fa3a81a3e36486c3a27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3777f97c1325bac0f8e76f25dbaef29ecf0887a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/3777f97c1325bac0f8e76f25dbaef29ecf0887a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
