)]}'
{"id":"openstack%2Fpuppet-cinder~943120","triplet_id":"openstack%2Fpuppet-cinder~master~I291f8592ca95636e16d73c1de71a378b36618c5a","project":"openstack/puppet-cinder","branch":"master","topic":"bug/2100712","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-03 11:03:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2025-03-03 03:17:23.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}},"hashtags":[],"change_id":"I291f8592ca95636e16d73c1de71a378b36618c5a","subject":"nfs: Ensure nfs client is installed","status":"MERGED","created":"2025-03-02 16:45:53.000000000","updated":"2025-03-03 14:51:09.000000000","submitted":"2025-03-03 14:50:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943120-bug/2100712","meta_rev_id":"2def1459dfd33b8ee185bd0736730cea16fa7511","_number":943120,"virtual_id_number":943120,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-03 14:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2025-03-03 03:17:23.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2025-03-03 09:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2025-03-03 09:16:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-02 18:32:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-02 23:51:57.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2025-03-03 03:17:23.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-03-03 09:16:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"173178d4687263d905f486b6b5b62885e6920b0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 16:45:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b066455e3804e61d9b05e0b61ddaa2dfdb5b5cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 16:46:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"baf6dd3e91cfd5beef449dfa21ea162fc6ece385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 18:32:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf07636c3c90488f8aeca344cfd11497\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db6e234166a54ad59ed45fbde3a71c5e : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bf5524c916c4979b076fb1255145f26 : SUCCESS in 2m 04s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9ab71f7ba544e469f990161012eaa94 : SUCCESS in 4m 41s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/b355e90e9f6949a79b3cff49327c5cc3 : FAILURE in 10m 35s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/648e89db20d341eaa03a098266a937b0 : FAILURE in 6m 38s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/af9903bcab624fe8bc0019f4a50a2934 : FAILURE in 7m 46s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c8d8638246348ae954253f54f6b9fad : FAILURE in 5m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7fdc6d6efb2440d8755e6d989b9ec74 : SUCCESS in 28m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86e674a214e7481ab08ed77712f9be5a : SUCCESS in 1h 02m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae2b0409e5804e49b3bbb58f7fb9e89a : SUCCESS in 32m 05s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ff690a388664c85a7e5a76d6d74e9c4 : SUCCESS in 30m 28s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bdaa5c5b8f842188d1598ca33c455de : FAILURE in 27m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7c353f9c5394957a816760afcedc5fc : SUCCESS in 1h 16m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fae2d3966caa4d698ec3a31ca0bd96d7 : SUCCESS in 1h 33m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2234fddba448407d8b80d1b04e5d3c12 : SUCCESS in 1h 16m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d70729f30ff4cdd91d4ad0bde921de9 : SUCCESS in 1h 08m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d078abe93484d2b8b1cfc86b1c6b600 : SUCCESS in 1h 08m 57s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9696022092fb4ec988ca25fd7b414bde : SUCCESS in 1h 19m 33s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af7308e41d2345428cdfa39eb5a37709 : SUCCESS in 1h 33m 22s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be889bb0bdbe4c09a4d56218ac319741 : SUCCESS in 40m 03s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abc6a95c2d3042dcb78ac98c49557a85 : SUCCESS in 1h 14m 58s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/75ea8465db824d8f9a3cc73c2f66e829 : SUCCESS in 54m 18s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f314ede46f4441bdb805b2dcf6db787f : SUCCESS in 24m 08s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98cb1cb2c0cd439b9d92b0bb4a628085 : SUCCESS in 12m 56s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1065650d1b224cf9a6c60053cd68f0a7 : SUCCESS in 29m 42s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d9ce8f9fb754a2598637fddf79cb455 : SUCCESS in 12m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07f9e72acb744048aa200f81bdebaffb : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0a9a486f1f62ee30f1daef0699795527e9033d4d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-02 23:51:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"945f9566ad1b06338682ab0a0b704017140d738a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 00:53:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4364fd7e293a5d29884f55644256dda6374621d1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 01:27:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"879872a54a54db219eaf79234bba184fd82f4ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 02:42:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c380311d28140e5bf2937fcf6493009\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/babeb1fffcc3401a813136668d342b5f : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/443027a0a12146dc806fd0c642839086 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2feefbc38164692af301752f4655f15 : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/21fb04f0989349e29e413550ed41e980 : SUCCESS in 13m 52s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ff42971b30d4d11ae17de6499d224ce : SUCCESS in 7m 26s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/84a7534652ee4cbda0a42035d33167d2 : SUCCESS in 10m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1204c606e89d40618f95b024fe5f6886 : SUCCESS in 6m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e73e1107cfeb46b490c2abfd642ff0e9 : SUCCESS in 50m 48s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94456d439c7c4a8d8a5d15380ed0bcec : FAILURE in 1h 17m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a86a62301864365915f58cffde82bbc : SUCCESS in 41m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ea18189d1904e3ab46bfca1a229b150 : SUCCESS in 43m 25s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ee9aac674ca428e91c12dc0de895909 : FAILURE in 23m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84ebb49d24394880a7baf890bf0b4afc : SUCCESS in 1h 11m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93dfecf09cf64aaba18ad29847183f5b : SUCCESS in 1h 25m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/73e8886cf5724230b85dac58b405cc76 : SUCCESS in 1h 03m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3630a3289fe4982a6856bdec986a8af : SUCCESS in 58m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fb5daa6b81049edb624ac583cea69e3 : SUCCESS in 1h 00m 21s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc646418d48147459dd287088d21c17d : SUCCESS in 1h 08m 13s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ac29b1e51f44006a38e4d81a105c4a3 : SUCCESS in 1h 18m 00s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c01c1bf864542e690be1649fc10c130 : SUCCESS in 1h 14m 52s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a74ced469dfe4f2a8cb188956a104e21 : SUCCESS in 59m 35s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/086a4c177bae4803b16e2889693bbfd2 : SUCCESS in 1h 16m 29s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/77007942a99e4136b21663895b373913 : SUCCESS in 34m 31s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f803f40498a2421ab906ed7a8470c42d : SUCCESS in 11m 18s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a11723ce2a64ad682992bea15b0454a : SUCCESS in 32m 46s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad60534d909843c8b8b2151950f00a22 : SUCCESS in 12m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da402983706c4e9cbe5ccf8853812211 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"8a469130063897c2927ef9700612dfd12614768a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-03 03:17:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5b0a75fb275248b932e24581b0817762cd09b747","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-03 09:16:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b4da12a29b0a62d8b75e89ff2a05bd49e1fd02b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 09:16:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c0ded5e3cae4b4ea6e298dcdec24ba7\n\n\nWarning:\n  Change 943119 in project openstack/puppet-manila does not share a change queue with 943120 in project openstack/puppet-cinder","accounts_in_message":[],"_revision_number":3},{"id":"dd4334742f8ecf2c1b3611da38a53126d8a3a3eb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 10:04:37.000000000","message":"Topic set to bug/2100712","accounts_in_message":[],"_revision_number":3},{"id":"05f5921b10f8e428a9d81d31b0755654f559372e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 11:03:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"88994f2bed7b77fe9758c2deb2ec839f59e1431f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 12:53:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc5f5127928049d08d68ad90033deac9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/409a118fe51e481688e0b9b7c00bf14b : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db1d0da10ccc46188a808943c9bebe22 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b27fc93591f049178c7ad3dc1e63016a : SUCCESS in 5m 41s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/acd6069679934697bd4049c020941df0 : SUCCESS in 10m 14s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/096f050543904e2a80b431ccd74fb8c8 : SUCCESS in 6m 45s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b43fef5775d4587b1d9bdd01f5ad5f1 : SUCCESS in 9m 57s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20a05be649c745248a1b7fe750c0ebcf : SUCCESS in 6m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b70669d178a44a139912deb2766ab918 : SUCCESS in 52m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29736a331d3a45ef9fa04c5083bda0ca : SUCCESS in 56m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b26f62762954bb181c291d0b40c75c6 : SUCCESS in 47m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f3c25461fbd4d1ca6334d4a19cef1d6 : SUCCESS in 37m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5de217fb38154d6eb1efdfd0defac9b7 : FAILURE in 32m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36e4553fe44e4438875842ed7f729388 : SUCCESS in 1h 28m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a1b7dad4f75464498461814521fd31b : SUCCESS in 1h 42m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/593ae1f45b22414c9df8bc3e47349dab : SUCCESS in 1h 20m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3ba01897968473b8b9c7c05a1a80958 : SUCCESS in 1h 22m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2533918efe84df78d7d2bf366be6305 : SUCCESS in 1h 15m 44s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd20d33805ed4f409f1439d67b56a2ee : SUCCESS in 1h 25m 41s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0130836df6a644699cb81b3372fbebef : SUCCESS in 1h 27m 27s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/76e1592bd59d4377a53cd70d3a61624b : SUCCESS in 1h 13m 56s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35b1f7b64fc0468086fcfaa68d5efe30 : SUCCESS in 1h 01m 04s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/65d632b3b9144f05a20eede7dcaf0ac8 : SUCCESS in 1h 16m 03s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3f28a2b93d24140918ca29103a47174 : SUCCESS in 26m 38s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9eb4e4dc0a4b4b67bdfb34a45b6e4378 : SUCCESS in 13m 19s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d48b635f854483daa500b57204970c3 : SUCCESS in 26m 19s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07874e07952e4c74b48bc9802796b522 : SUCCESS in 12m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aba97ecac4f461c924bbe9af78071b8 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7d61c460f97adb8b942f82807f729bb972489d44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 12:54:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1355d66a3155c6b1297a6bf94839163a28193d8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 14:50:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb15d2e89f647c4843064ad40becce5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b41b4848f4f74fc1a13b0808b78aa41d : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66ad7df2ae0140e3b644d82e8f87a727 : SUCCESS in 2m 39s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a96bad67f474e38868dd872b139e8c8 : SUCCESS in 9m 39s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1779a09daf284e5ba4dec7cc8aac4445 : SUCCESS in 7m 09s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/36297b7e03d548518b7e8574b7a09668 : SUCCESS in 10m 21s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bd3f78dfe0449b68f01343b9399cd99 : SUCCESS in 1h 22m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8031d607f43e4f779601acd7d7540a97 : SUCCESS in 1h 47m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54c453c9057f48ff94b7bd445ade664a : SUCCESS in 1h 10m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/835557a97ac0453b80be49311afd1f36 : SUCCESS in 1h 10m 38s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36179004c6d846878bacc9ed0864e150 : SUCCESS in 1h 11m 16s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ceeed949a9e476aba5aee10de04a5fa : SUCCESS in 29m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3b16fc3ee8f4d899718b8c34b55b1d4 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"4886c134b701c59118edd5c7a3b40b2addb63cf0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 14:50:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2def1459dfd33b8ee185bd0736730cea16fa7511","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 14:51:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6650f32fd4d44ae2b2d466ec273bc150\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3170629bfe2d4554bec3402ad1d4f21b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"75e727dea80289c36f37731db7e24270648f5438","revisions":{"f4efdffcdbb9b06fa1467aff680a72cc775b07c3":{"kind":"REWORK","_number":1,"created":"2025-03-02 16:45:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/943120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/20/943120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/1"}}},"commit":{"parents":[{"commit":"b5cbc9947b20b07ebb6be80e6eda92014fd499e5","subject":"Merge \"hpe3par: Replace redundant hard-coded default values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b5cbc9947b20b07ebb6be80e6eda92014fd499e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 16:44:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 16:44:28.000000000","tz":540},"subject":"nfs: Ensure nfs client is installed","message":"nfs: Ensure nfs client is installed\n\nThe nfs volume driver requires nfs client (especially the mount.nfs\ncommand installed by it). Ensure the package is installed.\n\nCloses-Bug: #2100712\nDepends-on: https://review.opendev.org/943119\nChange-Id: I291f8592ca95636e16d73c1de71a378b36618c5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/f4efdffcdbb9b06fa1467aff680a72cc775b07c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/f4efdffcdbb9b06fa1467aff680a72cc775b07c3"}]},"branch":"refs/heads/master"},"37a51b4125b2633bcdd0983bb87e14ab80681979":{"kind":"REWORK","_number":2,"created":"2025-03-02 16:46:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/943120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/20/943120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/2"}}},"commit":{"parents":[{"commit":"b5cbc9947b20b07ebb6be80e6eda92014fd499e5","subject":"Merge \"hpe3par: Replace redundant hard-coded default values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b5cbc9947b20b07ebb6be80e6eda92014fd499e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 16:44:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 16:46:06.000000000","tz":540},"subject":"nfs: Ensure nfs client is installed","message":"nfs: Ensure nfs client is installed\n\nThe nfs volume driver requires nfs client (especially the mount.nfs\ncommand installed by it). Ensure the package is installed.\n\nCloses-Bug: #2100712\nDepends-on: https://review.opendev.org/943119\nChange-Id: I291f8592ca95636e16d73c1de71a378b36618c5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/37a51b4125b2633bcdd0983bb87e14ab80681979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/37a51b4125b2633bcdd0983bb87e14ab80681979"}]},"branch":"refs/heads/master"},"75e727dea80289c36f37731db7e24270648f5438":{"kind":"REWORK","_number":3,"created":"2025-03-03 00:53:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/943120/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/20/943120/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/20/943120/3"}}},"commit":{"parents":[{"commit":"b5cbc9947b20b07ebb6be80e6eda92014fd499e5","subject":"Merge \"hpe3par: Replace redundant hard-coded default values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/b5cbc9947b20b07ebb6be80e6eda92014fd499e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 16:44:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-03 00:53:26.000000000","tz":540},"subject":"nfs: Ensure nfs client is installed","message":"nfs: Ensure nfs client is installed\n\nThe nfs volume driver requires nfs client (especially the mount.nfs\ncommand installed by it). Ensure the package is installed.\n\nCloses-Bug: #2100712\nDepends-on: https://review.opendev.org/943119\nChange-Id: I291f8592ca95636e16d73c1de71a378b36618c5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/75e727dea80289c36f37731db7e24270648f5438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/75e727dea80289c36f37731db7e24270648f5438"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
