)]}'
{"id":"openstack%2Fpuppet-cinder~957330","triplet_id":"openstack%2Fpuppet-cinder~master~I9b7bce7b18cef13fc20610ee2e5fa02721a50c42","project":"openstack/puppet-cinder","branch":"master","topic":"stdlib-func","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-15 08:00:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I9b7bce7b18cef13fc20610ee2e5fa02721a50c42","subject":"Switch to namespace sdtlib functions","status":"MERGED","created":"2025-08-14 09:14:49.000000000","updated":"2025-08-15 10:17:59.000000000","submitted":"2025-08-15 10:17:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957330-stdlib-func","meta_rev_id":"273f9e70b173dc9cc42d735acf56f2107b507cea","_number":957330,"virtual_id_number":957330,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-15 10:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-08-14 19:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-08-15 04:15:14.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-08-15 04:15:14.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 19:54:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-08-14 19:55:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-15 04:15:14.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"1f0d1b7c88f5b38b031d7c674b10caf0ed09d674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 09:14:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13c3da9c566ebaa8efd84a36f5956bc165e115fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 09:23:21.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d6ce67e240397146d882ddf2dbb4c379ccb48844","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 09:23:24.000000000","message":"Topic set to stdlib-func","accounts_in_message":[],"_revision_number":2},{"id":"2cc84dd3d7f3483927cdb337c0b8f66e960b68df","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-14 19:54:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7ffc671c6f058f41a42264ee449a63c2f450f979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 19:55:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190c2574be424634aeb3003a5f6ceab6\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/d1ebd00d997d4f0c95cf89ec31fc7bfc : SUCCESS in 2m 08s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72ea0be1533642bc861c6669ddc023ab : SUCCESS in 1m 26s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/591e1746c1ed40b19b6495f0d9c21b7c : SUCCESS in 1m 25s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/265cea83f1c44e02a9a93bd3ac762a09 : SUCCESS in 1m 25s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/6328d74b3a3c4ba1a9f0889d4b0b9a45 : SUCCESS in 5m 12s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e83d128ac2042d9b9691ccfb73e102c : SUCCESS in 5m 57s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb7b370a0cdf4b3b98d74ef56cd2d516 : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65cc6cad11564b3785d3f4ca0cdb9e9d : SUCCESS in 25m 49s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1375272609840f9af320c9e4fb9960f : SUCCESS in 1h 04m 53s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df215ebbfdb04bc5abc191bea9a6b7f7 : SUCCESS in 23m 26s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80a42a7c5a9d4173b6602fd75b78f45a : FAILURE in 42m 06s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80be6372f38b40d2b46951a073e65e01 : FAILURE in 43m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc817afaf6f94a6ab9761047ccc176a6 : SUCCESS in 1h 12m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/828953e4cef54d4db5881ae6e6c61408 : SUCCESS in 41m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4412f61e25cd4cb0a593ae12fa6daad1 : SUCCESS in 1h 03m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1642b59f7ee41fdbad3de60ee8719b0 : SUCCESS in 29m 11s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4999560939b14f1fa3bc275212071b90 : SUCCESS in 30m 32s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/88a4ed54d38f45c097ebc9f8dbfb3a39 : SUCCESS in 11m 08s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3cf5daf610e47d8a623027909633cf0 : SUCCESS in 11m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6348f5c03f84606bfa490f21f56b473 : SUCCESS in 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a7a1e85a30cba8b8e3e6ec79ca64ab280a4f8cc","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-15 04:15:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ea5587349080c267d6b111b4f6cba836f91646b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 04:15:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0759242578055d1f975110ebd81c0fdb8df73bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 05:42:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/271fa271c25841ac88f355d680da4ffc\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/dfd692429aa44fc4a68dd1dce724dd40 : SUCCESS in 3m 00s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba89b337055343608e4e82e3dee56596 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9bf22e145914034a75d59d1711d1664 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0027a9054fe94e82a9f27a6954c7fbec : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cb62f73420446b6a8ebc1f160b8319d : SUCCESS in 8m 19s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d47561b027ff4ef6846ae2a041c4cab2 : SUCCESS in 6m 49s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23eedd5869664d3ba7664aa0e532a0dc : FAILURE in 1h 08m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cb9b69ca2be4e738a8629913c274aac : SUCCESS in 1h 21m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3707fec69ccf42c0a18eb44ef3fa8096 : SUCCESS in 41m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d83ca9c93ed4dfd82cc6460e84516b8 : SUCCESS in 52m 59s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c484009d140a4d0ea9add20e80185205 : SUCCESS in 1h 00m 11s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d2ab4c39fe04c5980928a2450055d7f : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":2},{"id":"e79136fa9e437f6239caa5f88bc4d41bbe40ce5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-15 08:00:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d8382c38faa297c20d671e5db380da2327a978a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 09:12:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad4cf70f22e43fe906db381ad5dfad4\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/991b35598a2042d38c34b8d691f86363 : SUCCESS in 1m 39s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cf839aae8ba45119bfb43490699e137 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e065c98320b42c293bce90773b679f9 : SUCCESS in 1m 53s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddd9640b542d49489061c23150e7da07 : SUCCESS in 3m 14s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/730d53452166430c955459d452ffa450 : SUCCESS in 10m 20s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa1a3afb310b4d02a4786b336941a230 : SUCCESS in 3m 29s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64e694e24c6440de8e2af38496a87871 : SUCCESS in 6m 25s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f65043a4b4984127ab0fb2434725fd1e : SUCCESS in 29m 07s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b840b45dca9c408a94d88db18cdb204f : SUCCESS in 1h 09m 26s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ea34e50172d4928be3dc80f4befd457 : FAILURE in 25m 35s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ecd1f30fc9d495e9a5276732fef9d22 : FAILURE in 37m 24s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/021f735527e84ddabc54d905b9003731 : FAILURE in 34m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2844574c76db4a00bf2612fcb52524a6 : SUCCESS in 36m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b1fd889fcd44beb8c774023070a09d5 : SUCCESS in 50m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/158b1b5dbca0454486b8c157568756c0 : SUCCESS in 1h 01m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cfceb17e8ec477ba1fb92350b63dbf1 : SUCCESS in 32m 53s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec7058330de7445fb69c9eefca0e4e2f : SUCCESS in 57m 45s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/12a4812b69694e75963a79b3473296d0 : SUCCESS in 15m 39s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d984e404ff344ceba2b30fb59b310076 : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ecc339fa1a84f0da97b9c1768a47a5e : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a913a6580aaa0b030996691e7b4939ae875974b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 09:13:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ea89a6b74b821b0665adf38ff58eee06580eaff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 10:17:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d041f0a17c94e03a8ecaec244e0352a\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/dbde0caa7cab41d7ac18e52e6a99909e : SUCCESS in 2m 15s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a2918a1b94c4c97951a1c20ed0b4436 : SUCCESS in 2m 08s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dfbffc8387d4ba9ad0a5364831fe5c2 : SUCCESS in 1m 23s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3f3dd44d2cf409595d45b5c9f5190c3 : SUCCESS in 5m 06s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a09cc00ff1946f0b8f0272f46c80525 : SUCCESS in 12m 39s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a421c9eba8004eeab2e8f37dfc55b5a9 : SUCCESS in 4m 00s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4de3175493ca4b29bb1b36689cb1b03e : SUCCESS in 32m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b12bb3295f0403baa55d7867089d8dc : SUCCESS in 45m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fcf025a0f754746b8057c7db8a89f6b : SUCCESS in 1h 02m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64362e1cc25241479cc500d9095046c2 : SUCCESS in 59m 38s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f43cdf8fc0274596b2d5fd2365889a09 : SUCCESS in 1h 00m 12s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/9650f04831e049669da0189c1ce4a9de : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"273f9e70b173dc9cc42d735acf56f2107b507cea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 10:17:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"514fe5ce2604bfde59d666ca46a2e27c6527021f","revisions":{"f95b8ebb9aab335511aed461c3b6ac239dc3214d":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:14:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/957330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/30/957330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/1"}}},"commit":{"parents":[{"commit":"a1bfb682db5276d4f86ef7760f1b8a754016c50c","subject":"Add queue_manager and stream_fanout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/a1bfb682db5276d4f86ef7760f1b8a754016c50c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:14:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:14:42.000000000","tz":540},"subject":"witch to namespace sdtlib functions","message":"witch to namespace sdtlib functions\n\npuppetlabs-stdlib deprecated the existing functions in favor of\nthe new stdlib:: functions in 9.0.0. The version was released a few\nyears ago and now we assume we are ready to adopt to the migration.\n\nChange-Id: I9b7bce7b18cef13fc20610ee2e5fa02721a50c42\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/f95b8ebb9aab335511aed461c3b6ac239dc3214d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/f95b8ebb9aab335511aed461c3b6ac239dc3214d"}]},"branch":"refs/heads/master"},"514fe5ce2604bfde59d666ca46a2e27c6527021f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-14 09:23:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/957330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cinder","ref":"refs/changes/30/957330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cinder refs/changes/30/957330/2"}}},"commit":{"parents":[{"commit":"a1bfb682db5276d4f86ef7760f1b8a754016c50c","subject":"Add queue_manager and stream_fanout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/a1bfb682db5276d4f86ef7760f1b8a754016c50c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:14:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:23:21.000000000","tz":0},"subject":"Switch to namespace sdtlib functions","message":"Switch to namespace sdtlib functions\n\npuppetlabs-stdlib deprecated the existing functions in favor of\nthe new stdlib:: functions in 9.0.0. The version was released a few\nyears ago and now we assume we are ready to adopt to the migration.\n\nChange-Id: I9b7bce7b18cef13fc20610ee2e5fa02721a50c42\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/514fe5ce2604bfde59d666ca46a2e27c6527021f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cinder/commit/514fe5ce2604bfde59d666ca46a2e27c6527021f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
