)]}'
{"id":"openstack%2Fpuppet-cloudkitty~1004741","triplet_id":"openstack%2Fpuppet-cloudkitty~master~Ia07c1c1f547842cf0291d3e0f59e1bbe2a994ad2","project":"openstack/puppet-cloudkitty","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 16:11:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia07c1c1f547842cf0291d3e0f59e1bbe2a994ad2","subject":"Deprecate support for [database] mysql_enable_nbd","status":"MERGED","created":"2026-09-09 02:39:40.000000000","updated":"2026-09-10 16:13:12.000000000","submitted":"2026-09-10 16:11:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004741","meta_rev_id":"81f6aa7bae1de310c8c7880d9927be2f76d426f9","_number":1004741,"virtual_id_number":1004741,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 16:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2026-09-10 10:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-09-10 14:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2026-09-10 14:49:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 04:21:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 10:55:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-09-10 14:49:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a32bea7003c4dc089e7b02c59b3fcf1c8c928eb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:39:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"547506f802a5ef3ea835044a8789a1fd4d28cb9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:42:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8237123cbb20f318998b403d3fea32fd6ae32a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 04:21:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3642b342b6ef4877a17693a3681a4eea\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5bbb277255849bd8719f6eeeddf0157 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa44e8753ab7456cab6f73bc764e0d9f : SUCCESS in 3m 10s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/1cabbfb5065b4609869a8d9e93fe5d47 : FAILURE in 7m 55s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/829a7114c53d4f78b99698ae4fa12833 : FAILURE in 6m 11s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61c2966a4f6d4a4b83770b184d0194af : FAILURE in 5m 28s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb4e4c493e034eb6b823d3f50b36dc29 : SUCCESS in 1h 04m 06s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fa8357b759548489ae365a53f8fbea3 : SUCCESS in 55m 57s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/75baeadcc7fd4c999f4ce8e921a12d70 : SUCCESS in 34m 28s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f0582e45fff431989ee978d8f21acf2 : FAILURE in 6m 24s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecceb5ab2ba649bbaa06022c05152cdd : SUCCESS in 12m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/862284d7e06644b4a18ac105b1945fb5 : SUCCESS in 14m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8219acb2bb5d4b6392fd5cf1a2bac7d6 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4cb82d91d65c0d242dc9123db3975f312bc2d30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 07:35:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"65aa21f50136ee9dc2d3e880a1b8771955745848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 08:52:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e6bf4dad49f44ed9dab9ebcc84ee790\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b9dfb6e922b44f68151deee28a5f91d : SUCCESS in 3m 40s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c26e0bf1e78044e2a3d2b7baa30b9120 : SUCCESS in 2m 48s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1dd026beeb94defa04f482583c822bc : FAILURE in 4m 54s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc31a7b816b24d549aeb73286e8331ee : FAILURE in 5m 12s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f9fb93ffc544dfcb5b43b1831356ca3 : FAILURE in 4m 05s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45661c3147774eba88306a4ce1ade1c9 : SUCCESS in 59m 30s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2886aa475604e97bd2e1a8c7b0c83e4 : SUCCESS in 59m 01s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f2162d3a8bd48199e08502c5cd6f4a2 : SUCCESS in 1h 03m 40s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a323fca7f8d34f25a1c5ef9bf80ef755 : FAILURE in 8m 10s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2544558e4b944c483e67f2a60681900 : SUCCESS in 13m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73a43e1b29af4d67b197b14f652f4ca8 : SUCCESS in 10m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74457a4b416e49518116128f408f4354 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a041c3cdb4f844de6691553d968eeb52faf7cc7c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 16:18:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0056a0438776a046ccce82d120a1a0f0c3688b21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 17:29:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db1b1b52e784646930c935757ceb952\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecf1ec2edb3d4225b6026d620c8e294f : SUCCESS in 4m 33s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75467eb6ae2840149705616402e1375d : SUCCESS in 4m 21s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/8be68da22a9547eb8e69028f34f7fdbe : SUCCESS in 9m 24s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/872164be41be4189ab86c15cc2e052a3 : SUCCESS in 7m 32s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d86da4cd54d4b0d95c48a0e29686cd4 : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7adca0921b484434a095d684850f1781 : SUCCESS in 1h 05m 53s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25a9662aded647b3bd63adc9ca63970a : SUCCESS in 54m 22s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/16e9b6ceb5cc4fb1b254ba5cf3c362de : SUCCESS in 56m 03s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e65d107c203e4279a8a877bff7dfd5d0 : SUCCESS in 22m 55s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa73bb385b5745468cd47882485ee917 : SUCCESS in 19m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/819e19450df84a94ae77ac839f85830f : SUCCESS in 9m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28db6084c39742af9927e419e55ad483 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ff2a98c0ab4e3feba26c9fbe0c0ef2d8ae36b577","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-10 10:55:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1a812948bfcdad029171c6b4440623d7e9174dff","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-09-10 14:49:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a589e0cab5f55c307e3d5dd8282930b7fad222a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:54:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6564f76c47cf870517f617311abc95dd59923b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:11:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be2d2ea47e04339a0309a324f61610f\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c5d492150ec418ab4164c06989bf31c : SUCCESS in 3m 42s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4077e3dd296343939a578fc7feb4e3d1 : SUCCESS in 3m 15s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/aad6255a502f458a8961494b85bbac13 : SUCCESS in 6m 02s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b371a9f9b77483cb06dcf563fd64517 : SUCCESS in 8m 06s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/526509914f29477fb9498e9bb8359aa4 : SUCCESS in 1h 11m 20s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2dd69198ff141b38eac43804499016c : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01024fd8ef5a495a82ac2117b6c7d0be : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"88e5dc5d6b5b72ffc407b72cf882c3eed88f1e37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:11:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"81f6aa7bae1de310c8c7880d9927be2f76d426f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:13:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/995e7e35040940178a075ec5a0596260\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f0a2391de984b7b8349926b025462cc : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0531b0b182709df55ac41b495bea0f25c0ae583f","revisions":{"660046a57b9181125dd917257a42c45264fb4d15":{"kind":"REWORK","_number":1,"created":"2026-09-09 02:39:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/1004741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cloudkitty","ref":"refs/changes/41/1004741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/1"}}},"commit":{"parents":[{"commit":"fa71a8427297b70f8a15488bc2885378dfec6505","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/fa71a8427297b70f8a15488bc2885378dfec6505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:39:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:39:30.000000000","tz":540},"subject":"Deprecate support for [database] mysql_enable_nbd","message":"Deprecate support for [database] mysql_enable_nbd\n\nDepends-on: https://review.opendev.org/1004735\nChange-Id: Ia07c1c1f547842cf0291d3e0f59e1bbe2a994ad2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/660046a57b9181125dd917257a42c45264fb4d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/660046a57b9181125dd917257a42c45264fb4d15"}]},"branch":"refs/heads/master"},"279cd6c608a6b5c1e0ecc0fd537e84776d69ca16":{"kind":"REWORK","_number":2,"created":"2026-09-09 02:42:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/1004741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cloudkitty","ref":"refs/changes/41/1004741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/2"}}},"commit":{"parents":[{"commit":"fa71a8427297b70f8a15488bc2885378dfec6505","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/fa71a8427297b70f8a15488bc2885378dfec6505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:39:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:42:23.000000000","tz":540},"subject":"Deprecate support for [database] mysql_enable_nbd","message":"Deprecate support for [database] mysql_enable_nbd\n\nDepends-on: https://review.opendev.org/1004735\nChange-Id: Ia07c1c1f547842cf0291d3e0f59e1bbe2a994ad2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/279cd6c608a6b5c1e0ecc0fd537e84776d69ca16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/279cd6c608a6b5c1e0ecc0fd537e84776d69ca16"}]},"branch":"refs/heads/master"},"0531b0b182709df55ac41b495bea0f25c0ae583f":{"kind":"REWORK","_number":3,"created":"2026-09-09 07:35:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/1004741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-cloudkitty","ref":"refs/changes/41/1004741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-cloudkitty refs/changes/41/1004741/3"}}},"commit":{"parents":[{"commit":"fa71a8427297b70f8a15488bc2885378dfec6505","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/fa71a8427297b70f8a15488bc2885378dfec6505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:39:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 07:35:33.000000000","tz":540},"subject":"Deprecate support for [database] mysql_enable_nbd","message":"Deprecate support for [database] mysql_enable_nbd\n\nDepends-on: https://review.opendev.org/1004735\nChange-Id: Ia07c1c1f547842cf0291d3e0f59e1bbe2a994ad2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/0531b0b182709df55ac41b495bea0f25c0ae583f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-cloudkitty/commit/0531b0b182709df55ac41b495bea0f25c0ae583f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
