)]}'
{"id":"openstack%2Fpuppet-congress~656129","triplet_id":"openstack%2Fpuppet-congress~master~I84c45109e8b8b6767c9ede3f5a854e99eba83b30","project":"openstack/puppet-congress","branch":"master","topic":"bug/1826692","hashtags":[],"change_id":"I84c45109e8b8b6767c9ede3f5a854e99eba83b30","subject":"Deprecate idle_timeout option","status":"MERGED","created":"2019-04-28 07:01:01.000000000","updated":"2019-04-29 07:54:15.000000000","submitted":"2019-04-29 07:54:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656129-1556524455204-7d7fc31e","meta_rev_id":"f605875b37e55f72f35a6cc0dd25b779c396941c","_number":656129,"virtual_id_number":656129,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-04-29 07:05:52.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-29 07:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-04-28 08:42:01.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-28 17:16:37.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-04-28 17:16:37.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 08:42:01.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-28 17:16:37.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-29 07:54:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c19289f7415419dc85b3c42f56f868bac1d8dde4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 07:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"372eb98e2c9d4028a9066cef04542c2a218f9a80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 07:17:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-lint-ubuntu-bionic/8cc3d03/ : SUCCESS in 4m 44s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-syntax-4-ubuntu-bionic/912a52a/ : SUCCESS in 5m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-syntax-5-ubuntu-bionic/6eacea5/ : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-syntax-6-ubuntu-bionic/10b3557/ : SUCCESS in 5m 13s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-4.8-centos-7/dbb1699/ : SUCCESS in 10m 31s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-5.5-centos-7/3125462/ : SUCCESS in 9m 39s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/7eef4e2/ : SUCCESS in 8m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-latest-ubuntu-bionic/879f909/ : SUCCESS in 7m 52s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/29/656129/1/check/build-openstack-releasenotes/7661fe4/html/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1363c63fdb4805a2cc2cbf36863d130635729db1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-28 08:42:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49928626d57d57d69951a259588c7806f65c03bd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-28 17:16:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfb8b5f1f5b733e88fd9052083f8447594e9c701","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-29 07:05:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4657378e6639c19149cb3737e5556cf7ba4380e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:30:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-lint-ubuntu-bionic/7db39bd/ : SUCCESS in 4m 27s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-syntax-4-ubuntu-bionic/47636f0/ : SUCCESS in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-syntax-5-ubuntu-bionic/2b6448c/ : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-syntax-6-ubuntu-bionic/11d47df/ : SUCCESS in 4m 36s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-4.8-centos-7/c927328/ : SUCCESS in 13m 30s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-5.5-centos-7/e063178/ : SUCCESS in 14m 10s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/9f512fa/ : SUCCESS in 10m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/656129/1/check/puppet-openstack-unit-latest-ubuntu-bionic/21c7f08/ : SUCCESS in 6m 38s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/29/656129/1/check/build-openstack-releasenotes/482cea6/html/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d281263df906cc08a6d2c2b641bfe9e6754f2dbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:31:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"830d3beed3d3c2b1113bb0d7ffb93a2b1829584d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:54:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/656129/1/gate/puppet-openstack-lint-ubuntu-bionic/d523ff5/ : SUCCESS in 4m 36s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/656129/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/d9b0daa/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/656129/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/1837390/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/656129/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/56de6a1/ : SUCCESS in 4m 09s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/656129/1/gate/puppet-openstack-unit-4.8-centos-7/37c011d/ : SUCCESS in 11m 05s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/656129/1/gate/puppet-openstack-unit-5.5-centos-7/f86b10a/ : SUCCESS in 9m 17s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/656129/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/8be5ec1/ : SUCCESS in 7m 44s\n- build-openstack-releasenotes http://logs.openstack.org/29/656129/1/gate/build-openstack-releasenotes/1b3f0f6/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ed6cdff8dc9d81a5bd0fe28864064d76f3b5016e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:54:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f2c3175596ee5e9b811eea32c54a0edebdc6efc","revisions":{"1f2c3175596ee5e9b811eea32c54a0edebdc6efc":{"kind":"REWORK","_number":1,"created":"2019-04-28 07:01:01.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/29/656129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-congress","ref":"refs/changes/29/656129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-congress refs/changes/29/656129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-congress refs/changes/29/656129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-congress refs/changes/29/656129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-congress refs/changes/29/656129/1"}}},"commit":{"parents":[{"commit":"c2eea48883fab0c96c187bc6c6a753c1305b9e5f","subject":"Move to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-congress/commit/c2eea48883fab0c96c187bc6c6a753c1305b9e5f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:07.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:07.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: I84c45109e8b8b6767c9ede3f5a854e99eba83b30\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-congress/commit/1f2c3175596ee5e9b811eea32c54a0edebdc6efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-congress/commit/1f2c3175596ee5e9b811eea32c54a0edebdc6efc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
