)]}'
{"id":"openstack%2Fpuppet-congress~684476","triplet_id":"openstack%2Fpuppet-congress~master~Ib1c7c23a28d99451d30533124ff69f37e4e8333c","project":"openstack/puppet-congress","branch":"master","topic":"prep-release-train-rc1","hashtags":[],"change_id":"Ib1c7c23a28d99451d30533124ff69f37e4e8333c","subject":"Prepare Train RC1","status":"MERGED","created":"2019-09-25 05:46:52.000000000","updated":"2019-09-27 01:58:11.000000000","submitted":"2019-09-27 01:58:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684476-1569549491967-91403195","meta_rev_id":"37f3fdd0e4a25d22355d873bedfc019864022553","_number":684476,"virtual_id_number":684476,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 01:58:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:47:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:47:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:47:08.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 01:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"145d4fe88efd78e6234d57bbb02f30b6d9dd5b68","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-25 05:46:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a44bf47a9e0da62581ed773e1949afa5461450d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 05:56:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/8e238b0e08c04b2b86ad27aebaa6155a : SUCCESS in 1m 59s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a358b864b7a544828011f8ce1500a38b : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6d0e622c6446a75ae2be854db4db89004b14aa36","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:47:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b013ccd7e497615e021c81b3624c944bcc091f5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:47:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19ce54d997103e39f334a907a872ca0e75635028","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:58:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c7a2bd38168e863b6e5139083b2d2deb458a6dfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:58:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/134845d74af6485d9b7192ba4eb9cada : SUCCESS in 3m 22s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7dfde4fe2604e1388e148b8bd51ce13 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"229d549109bc4ed50284f0fd98551d69f2cb2e57","revisions":{"229d549109bc4ed50284f0fd98551d69f2cb2e57":{"kind":"REWORK","_number":1,"created":"2019-09-25 05:46:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/76/684476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-congress","ref":"refs/changes/76/684476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-congress refs/changes/76/684476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-congress refs/changes/76/684476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-congress refs/changes/76/684476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-congress refs/changes/76/684476/1"}}},"commit":{"parents":[{"commit":"2a7e2c72bcfed9ebee23c58bf8ac9ba6f356cfd4","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-congress/commit/2a7e2c72bcfed9ebee23c58bf8ac9ba6f356cfd4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:42.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:42.000000000","tz":480},"subject":"Prepare Train RC1","message":"Prepare Train RC1\n\nUpdate the version metadata for Train milestone RC1\n\nChange-Id: Ib1c7c23a28d99451d30533124ff69f37e4e8333c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-congress/commit/229d549109bc4ed50284f0fd98551d69f2cb2e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-congress/commit/229d549109bc4ed50284f0fd98551d69f2cb2e57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
