)]}'
{"id":"openstack%2Fpuppet-designate~665784","triplet_id":"openstack%2Fpuppet-designate~master~I6d05a9130e5f7bbe850c6ac04553a844de24b574","project":"openstack/puppet-designate","branch":"master","hashtags":[],"change_id":"I6d05a9130e5f7bbe850c6ac04553a844de24b574","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:14:52.000000000","updated":"2020-09-18 22:55:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0a04d4ec27ddfa8283043c904ba12dfd80d6e91a","_number":665784,"virtual_id_number":665784,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-17 23:51:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 23:51:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"493f5f6d6d9e6b2e2f50761b0b4a7395f95332ba","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:14:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c56e64d5d6f24475284c2e085622a7eb24e50b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 23:51:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/665784/1/check/puppet-openstack-lint-ubuntu-bionic/ed4929b/ : SUCCESS in 6m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/665784/1/check/puppet-openstack-syntax-5-ubuntu-bionic/00539f4/ : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/665784/1/check/puppet-openstack-syntax-6-ubuntu-bionic/f46d866/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/665784/1/check/puppet-openstack-unit-5.5-centos-7/0a3291f/ : FAILURE in 11m 57s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/84/665784/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/0cc5d9e/ : FAILURE in 12m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/665784/1/check/puppet-openstack-unit-latest-ubuntu-bionic/b7d0713/ : FAILURE in 8m 31s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/84/665784/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/01de1d5/ : FAILURE in 4m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/84/665784/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/1ef0cfd/ : FAILURE in 14m 59s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/84/665784/1/check/puppet-openstack-beaker-centos-7/e42122a/ : FAILURE in 9m 43s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/84/665784/1/check/puppet-openstack-beaker-ubuntu-bionic/d2b4bfd/ : FAILURE in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a04d4ec27ddfa8283043c904ba12dfd80d6e91a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:55:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af06b726de7c89627c36f90fbd254aaaa7f103c3","revisions":{"af06b726de7c89627c36f90fbd254aaaa7f103c3":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:14:52.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/84/665784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/84/665784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/84/665784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/84/665784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/84/665784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/84/665784/1"}}},"commit":{"parents":[{"commit":"7936ed1388115c155072bcd4235e329e538120cb","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/7936ed1388115c155072bcd4235e329e538120cb"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:14:43.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:14:43.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I6d05a9130e5f7bbe850c6ac04553a844de24b574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/af06b726de7c89627c36f90fbd254aaaa7f103c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/af06b726de7c89627c36f90fbd254aaaa7f103c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
