)]}'
{"id":"openstack%2Fpuppet-designate~677627","triplet_id":"openstack%2Fpuppet-designate~master~I3e4d9da942b352d446613c7241539f564b753d64","project":"openstack/puppet-designate","branch":"master","topic":"bug/1840868","hashtags":[],"change_id":"I3e4d9da942b352d446613c7241539f564b753d64","subject":"Introduce the new rabbit_heartbeat_in_pthread option","status":"MERGED","created":"2019-08-21 06:25:18.000000000","updated":"2019-08-22 03:30:43.000000000","submitted":"2019-08-22 03:30:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677627-1566444643113-0f3fa659","meta_rev_id":"23c9f63e9f32bcadc3d0aed961a426b5daaedd0e","_number":677627,"virtual_id_number":677627,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-22 03:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 02:23:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-21 15:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-22 02:23:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 15:16:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-22 02:23:38.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 03:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8db2bb25882c31c46824c8f8016a6f5136c6efd4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-21 06:25:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b071d103d4def83b9be1c1937f01910225f4d08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:47:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37c47809aec44e1dbadddf3d174453c5 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0565c5156c8459d8260bd9fcbe404a1 : SUCCESS in 6m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a5c52c926f64dd6889aeda5cbe43d29 : SUCCESS in 6m 32s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ec23bbd7e8c948e1a7e3d821be388245 : SUCCESS in 20m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dea8e0861d284aa4a58e30a02e6a2f8d : SUCCESS in 14m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad4b566c2a3b4bbc8dc34a68a96c05f3 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6dff23b29c34859b7e4f1c2cf6b3ed6 : FAILURE in 32m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2f701bfa89324cd6baea6a586317ccb1 : SUCCESS in 1h 17m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4fca3288e03b49f9b2eeaecbd53a5890 : SUCCESS in 20m 24s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b91758cb5573456ca642ea3cae3bc01c : FAILURE in 8m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97be4ded85f747798f1936b397c09c24 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"16be381a0675689741899389a227fbfc689ff758","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-21 15:16:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"841d843a327365c922805ad524e1278ad6bb92b3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:23:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0359625204107da4c4532da4d3521fe216336e86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 02:23:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1642dc856aa597d2e5c7db0584c4f57a7f95db05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:30:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54f31ea353af45b997219bee8a56a525 : SUCCESS in 7m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be2eee1c69c940c893f642cf0bc3afaa : SUCCESS in 5m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b59548cbca524f2aab6dd111c56aecdc : SUCCESS in 4m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/21727d36e36f40e196cf727a34211d22 : SUCCESS in 18m 56s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/002aeea4ed1c4328ab76aa5076a89ccb : SUCCESS in 13m 13s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/93eb38ea7cf4403580c6db5e7af0780b : SUCCESS in 1h 01m 37s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/66d72410929a4df0b7a8854ce6412851 : SUCCESS in 17m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2f9fa01ba8c4197b41a410cc6ffa375 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"57e980cffb6a1bddd92e79c4bb27d423a348a3fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:30:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"521337906e9e4677340ebf9756bc3f2c76a65180","revisions":{"521337906e9e4677340ebf9756bc3f2c76a65180":{"kind":"REWORK","_number":1,"created":"2019-08-21 06:25:18.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/27/677627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/27/677627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/677627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/677627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/677627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/27/677627/1"}}},"commit":{"parents":[{"commit":"d4c44694e58a7e45d3d2707fcd19a94c6a43b7aa","subject":"Remove deprecated pki related options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d4c44694e58a7e45d3d2707fcd19a94c6a43b7aa"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:21:01.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:23:26.000000000","tz":480},"subject":"Introduce the new rabbit_heartbeat_in_pthread option","message":"Introduce the new rabbit_heartbeat_in_pthread option\n\noslo.messaging RabbitMQ driver have now a new option that allow user to\nrun the RabbitMQ heartbeat over a native python thread.\n\nThese change allow user to use this new option.\n\nChange-Id: I3e4d9da942b352d446613c7241539f564b753d64\nCloses-Bug: #1840868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/521337906e9e4677340ebf9756bc3f2c76a65180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/521337906e9e4677340ebf9756bc3f2c76a65180"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
