)]}'
{"id":"openstack%2Fpuppet-designate~687333","triplet_id":"openstack%2Fpuppet-designate~stable%2Ftrain~Ic9ca6ee2e7cf927d926155ea73e65fb250ca083a","project":"openstack/puppet-designate","branch":"stable/train","topic":"deprecate-pool-manager","hashtags":[],"change_id":"Ic9ca6ee2e7cf927d926155ea73e65fb250ca083a","subject":"Deprecated designate::pool_manager","status":"MERGED","created":"2019-10-08 15:33:53.000000000","updated":"2019-10-09 18:01:42.000000000","submitted":"2019-10-09 17:59:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687333-1570643996425-33a7a8e1","meta_rev_id":"19750b343869b3f3ff863139a07fb41b72b6450d","_number":687333,"virtual_id_number":687333,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-10-09 15:28:42.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 17:59:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-09 12:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-10-08 17:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"date":"2019-10-09 08:03:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-10-09 12:51:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 17:59:36.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-09 08:03:45.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-09 12:51:16.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-10-09 17:59:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59af2d6c42a96f47de2f9695e1f73424e6b84292","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 15:33:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"763e1467a8de65004aa4cfb6ce6c19baf1e76164","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 15:35:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9b22516543a5c06de3540ece33b5a3f0f69ddf42","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 16:07:02.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3e51bddfdd4d1f7fd9991b7eba0d7a9bed77da8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 17:24:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6824d7810caf4cd983dfac1bfda8134d : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/091cf8abfa6e463386405a5b7df11ee7 : SUCCESS in 5m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/44538bd408da46bb950feaf13afe3b0f : SUCCESS in 12m 28s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd664e5623494207963d148517c8982a : FAILURE in 29m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5a351e5804f3478a87cee23bcd1ca78c : SUCCESS in 1h 07m 34s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/60cf1468e1d748829707c944e95d3c1d : SUCCESS in 19m 57s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ad880bd372c468a95bdb8f47762e1d6 : FAILURE in 13m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0052a180336d433fb9bcdc5eaf72bf44 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ce683fe195974d9a408fa42e6d548146d4bf3662","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-08 17:59:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"aacb6bed642fa6d85e9e6a0efaf82c1ec834dd3b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-09 12:51:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0bd2a5a8f11b86bf2cb1e2e2805257dc0bcde7ae","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-09 13:28:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"62f1e512b94e7814647a6098958a8d8ee12c3551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:07:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a425fdec632f4615821541e98fc63c09 : SUCCESS in 6m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5bafe91e10d46d69944d838ed3337c0 : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/21530a060f174becaeb92cde6707da42 : SUCCESS in 16m 18s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efeef749469649aa89ade31fa17535a6 : FAILURE in 26m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a70dc6097fc840bb8993149357f0fd79 : FAILURE in 1h 37m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/01721b73a6704816bcdf824d6d3a2bf4 : SUCCESS in 15m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05bf459b98e94226a6531ccef583f639 : FAILURE in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21513524f8474f819293f82c3db5b036 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"38a88c89977aa648ae3d5064d474b12bf8eb4de5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-09 15:28:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4f5959bf50dd3966a09bb800df0e0413f9a97501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:43:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c85883585e084927815e2007e8725b3f : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81ee0d91f97b4f7eabb92623d294b454 : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/320038fd4b704edabed1ec085b748602 : SUCCESS in 12m 52s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1073bbad39441f1894e713d6df1f829 : FAILURE in 27m 33s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d6f0abe32ed841dd98dd383c9ef36637 : SUCCESS in 1h 12m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e0d0c6f771164f549fe2813dd06d4eba : SUCCESS in 21m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bc4090d740d461b90e7a5bb272f77dc : FAILURE in 8m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe1315800b0d468284cfcdcd4e0ed2f9 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"6eb7837e3305d6a4bf76d5e5d45db1d7033ec2df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:44:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f47a43527a5b484a753d888d6362e5c04ac4173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 17:59:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b77f48a98b54b38bed21996e35fa588 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e6743e3f81c479f84757da6cb62d2ff : SUCCESS in 3m 44s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b3ff1e615a7248fbad98526b5bb8262d : SUCCESS in 18m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b9d44693d389485481fec4a2b770562d : SUCCESS in 1h 09m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1b8c5d6b83e14b59a2cab180a28b9d72 : SUCCESS in 17m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24816b2fec014f43918d896bc790e10a : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"37c6a2cb28b5e17afaf1e9a49d4bce95512b4edf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 17:59:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a430043893d1feb2fbec534438c27dd0a1b6209a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:01:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e9fd9c9c9354fca8a8ed55e90a9f779 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a89d6bc336ae4f87a48ecbc915bbd4155582eaf7","revisions":{"24c65de9070c2f4cb47081bc376b67550313917d":{"kind":"REWORK","_number":1,"created":"2019-10-08 15:33:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/33/687333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/33/687333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/1"}}},"commit":{"parents":[{"commit":"77a225daeabd6e48a6878baa6faac76d2dab3710","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/77a225daeabd6e48a6878baa6faac76d2dab3710"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:32:11.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:32:11.000000000","tz":120},"subject":"Deprecated designate::pool_manager","message":"Deprecated designate::pool_manager\n\nThe pool-manager is completely removed from\nDesignate in Train [1] so we should deprecate\nit before final release.\n\n[1] https://docs.openstack.org/releasenotes/designate/train.html\n\nChange-Id: Ic9ca6ee2e7cf927d926155ea73e65fb250ca083a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/24c65de9070c2f4cb47081bc376b67550313917d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/24c65de9070c2f4cb47081bc376b67550313917d"}]},"branch":"refs/heads/stable/train"},"03dcb5a7864592ddb33603eefa12c5d1bbfaad6a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-08 15:35:18.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/33/687333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/33/687333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/2"}}},"commit":{"parents":[{"commit":"77a225daeabd6e48a6878baa6faac76d2dab3710","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/77a225daeabd6e48a6878baa6faac76d2dab3710"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:32:11.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:35:16.000000000","tz":0},"subject":"Deprecated designate::pool_manager","message":"Deprecated designate::pool_manager\n\nThe pool-manager is completely removed from\nDesignate in Train [1] so we should deprecate\nit before final release.\n\n[1] https://docs.openstack.org/releasenotes/designate/train.html\n\nDepends-On: https://review.opendev.org/#/c/687247/\nChange-Id: Ic9ca6ee2e7cf927d926155ea73e65fb250ca083a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/03dcb5a7864592ddb33603eefa12c5d1bbfaad6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/03dcb5a7864592ddb33603eefa12c5d1bbfaad6a"}]},"branch":"refs/heads/stable/train"},"a89d6bc336ae4f87a48ecbc915bbd4155582eaf7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-08 16:07:02.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/33/687333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/33/687333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/33/687333/3"}}},"commit":{"parents":[{"commit":"77a225daeabd6e48a6878baa6faac76d2dab3710","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/77a225daeabd6e48a6878baa6faac76d2dab3710"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:32:11.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 16:07:00.000000000","tz":0},"subject":"Deprecated designate::pool_manager","message":"Deprecated designate::pool_manager\n\nThe pool-manager is completely removed from\nDesignate in Train [1] so we should deprecate\nit before final release.\n\n[1] https://docs.openstack.org/releasenotes/designate/train.html\n\nDepends-On: https://review.opendev.org/#/c/687343/\nChange-Id: Ic9ca6ee2e7cf927d926155ea73e65fb250ca083a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a89d6bc336ae4f87a48ecbc915bbd4155582eaf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a89d6bc336ae4f87a48ecbc915bbd4155582eaf7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
