)]}'
{"id":"openstack%2Fpuppet-designate~687334","triplet_id":"openstack%2Fpuppet-designate~stable%2Ftrain~Ie69c64795803c84d41d0a31b7e0a29ea493781eb","project":"openstack/puppet-designate","branch":"stable/train","topic":"release","hashtags":[],"change_id":"Ie69c64795803c84d41d0a31b7e0a29ea493781eb","subject":"Release 15.5.0","status":"MERGED","created":"2019-10-08 15:34:28.000000000","updated":"2019-10-09 17:59:57.000000000","submitted":"2019-10-09 17:59:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687334-1570643997419-bad3269c","meta_rev_id":"e8c952e708140defa8845c244b404a358fd0b3d7","_number":687334,"virtual_id_number":687334,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 17:59:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-10-09 06:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-10-08 17:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-10-09 06:14:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 17:59:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-09 06:14:53.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-10-09 17:59:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc55001c75a5fb7890f2e7a289f55a47ac85be08","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 15:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"561f6e0820893d352c5afb7c51cc5a9ac57470c1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 15:35:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"23f5931536cacbaa47f3f592b3219c9f71d84a1d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 15:40:42.000000000","message":"Patch Set 2:\n\nSee https://review.opendev.org/687337 for pending openstack/releases patch","accounts_in_message":[],"_revision_number":2},{"id":"eefbe887b53538c66db108964a6fee018e3fcc14","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 16:07:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7570bcf11f25ba2a76fe0a00ef8bf67d6c6c2ac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 16:23:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c5513172dec7446ea7e39d1745a38b25 : SUCCESS in 2m 07s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fbb32268a3e43788e1ec98fd401d094 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"64fb574adefc7c7bc81cf3bcb3f91a6e3f9abadc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-08 17:59:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e40ebf562c81499917011f0a79e5b4f6421b96f1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-09 06:14:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ff0141c8084aa615643dc3f7463d34a7ab2fbade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:44:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cfbcda4413350f4e59e127a75071ba6143396871","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 17:59:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d3ab343dd1ee7aef9d21d4d9730879b31ece4269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 17:59:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b04913236d2745cfb4bdd862e90cbf93 : SUCCESS in 3m 47s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1593bd1e632846c4817b304e72d2f115 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0543a1c2a0108fb7b67c41e6bc2fbad99eb24cb9","revisions":{"ab13ba894cea4ff9fb77f4ef10741aeb6bf2d90a":{"kind":"REWORK","_number":1,"created":"2019-10-08 15:34:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/34/687334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/34/687334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/1"}}},"commit":{"parents":[{"commit":"24c65de9070c2f4cb47081bc376b67550313917d","subject":"Deprecated designate::pool_manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/24c65de9070c2f4cb47081bc376b67550313917d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:34:09.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:34:09.000000000","tz":120},"subject":"Release 15.5.0","message":"Release 15.5.0\n\nChange-Id: Ie69c64795803c84d41d0a31b7e0a29ea493781eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/ab13ba894cea4ff9fb77f4ef10741aeb6bf2d90a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/ab13ba894cea4ff9fb77f4ef10741aeb6bf2d90a"}]},"branch":"refs/heads/stable/train"},"897a90d0d74904514700f2f92e8a48e3eb57c3d6":{"kind":"NO_CHANGE","_number":2,"created":"2019-10-08 15:35:25.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/34/687334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/34/687334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/2"}}},"commit":{"parents":[{"commit":"03dcb5a7864592ddb33603eefa12c5d1bbfaad6a","subject":"Deprecated designate::pool_manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/03dcb5a7864592ddb33603eefa12c5d1bbfaad6a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:34:09.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:35:25.000000000","tz":0},"subject":"Release 15.5.0","message":"Release 15.5.0\n\nChange-Id: Ie69c64795803c84d41d0a31b7e0a29ea493781eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/897a90d0d74904514700f2f92e8a48e3eb57c3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/897a90d0d74904514700f2f92e8a48e3eb57c3d6"}]},"branch":"refs/heads/stable/train"},"0543a1c2a0108fb7b67c41e6bc2fbad99eb24cb9":{"kind":"NO_CHANGE","_number":3,"created":"2019-10-08 16:07:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/34/687334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/34/687334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/34/687334/3"}}},"commit":{"parents":[{"commit":"a89d6bc336ae4f87a48ecbc915bbd4155582eaf7","subject":"Deprecated designate::pool_manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a89d6bc336ae4f87a48ecbc915bbd4155582eaf7"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:34:09.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 16:07:11.000000000","tz":0},"subject":"Release 15.5.0","message":"Release 15.5.0\n\nChange-Id: Ie69c64795803c84d41d0a31b7e0a29ea493781eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0543a1c2a0108fb7b67c41e6bc2fbad99eb24cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0543a1c2a0108fb7b67c41e6bc2fbad99eb24cb9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
