)]}'
{"id":"openstack%2Fpuppet-designate~723100","triplet_id":"openstack%2Fpuppet-designate~master~If22e4f75a0959507d6d3a97036f873df2f284d03","project":"openstack/puppet-designate","branch":"master","topic":"deprecate-worker-notify","hashtags":[],"change_id":"If22e4f75a0959507d6d3a97036f873df2f284d03","subject":"Deprecate designate::worker::worker_notify","status":"MERGED","created":"2020-04-25 14:29:58.000000000","updated":"2020-04-27 19:49:01.000000000","submitted":"2020-04-27 19:47:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723100-1588016841945-54f1cd87","meta_rev_id":"d7c46bd2b9d94ca7828b4c6755a144b1af7bf2d8","_number":723100,"virtual_id_number":723100,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-25 15:41:27.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 19:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-26 14:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-04-27 14:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-04-27 14:56:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 14:47:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-27 14:56:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-27 19:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1731c19e8d3fabb8316c3f4ac30c6f3f81916d1d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-25 14:29:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6561162543523ec3631513829433d7e6141a4ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-25 14:49:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb6512390974be028f0d1959741a3b8d2de294a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 15:38:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/427e1f831a954b1bb268c3791b64081b : SUCCESS in 5m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b199911e8fa479f86c9da5a666c053e : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/824958cf1bcb4058b5e2bbc759d425ad : SUCCESS in 5m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e7425e38a8f84d6787d4032db42b30f4 : SUCCESS in 16m 22s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c5aec2b2b7ec4cc083cec92e71ab6811 : SUCCESS in 13m 19s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74cf742abf93438ebc4fdba3c4a0becf : SUCCESS in 9m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdeb31c8d5de41c297567cda4d9ec3d8 : SUCCESS in 17m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77954440215345948760e9d61b3f05f8 : SUCCESS in 42m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c8e7a32b87a94018a9592b31f805e989 : EXCEPTION\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/38315a28cbf94e80826391882967061e : SUCCESS in 19m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7464fc699fab4e6786d9b2577028aca0 : FAILURE in 13m 20s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34b0bbd7457f4ad1a78510b3569cb956 : FAILURE in 7m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed44398da7654ed8975ae87b5e1745ce : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"efb31aa40a3f37428971be8a108840365f480851","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-25 15:41:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"df641fe32c22fe427bdf11605c38c237e81b7798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 17:19:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39e66d901e0f4e19a15c3e31736a7fd3 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9c0d49385664123b0b1e413f48befbd : SUCCESS in 5m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e581ffaaad9e40118b29017bd26c21e7 : SUCCESS in 7m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/adcb27dbd1694a159ace33c2d48948a2 : SUCCESS in 10m 43s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/24d4fe92dc7245e281c7d196ca39414d : SUCCESS in 16m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76e33052993b45c3ac3f0ed8372ce342 : SUCCESS in 13m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75927a3844f747a3a6536e32cba60870 : SUCCESS in 11m 03s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a05a75478c744d4ba8c1a5cde970c57b : SUCCESS in 40m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e4ec17fc75f4b23b3c04a70c82bf04f : SUCCESS in 1h 11m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ef989162231b4cddab4f0093338c3bc1 : SUCCESS in 20m 38s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e1ea66815acd4224ba9c938b0d47c9f7 : FAILURE in 8m 38s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9458e99ca4664be49aca0289b0d02d3a : FAILURE in 11m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f2d61634cb742d5a5247f9ef1196b40 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7f14455178d7f9a11ed50e05f6dbe01af3c64fa8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-26 14:47:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eb066b1dad3a048bece2df91866e8e228727c6b4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-27 14:56:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"582c30c606764b040cf70996fda5bf7e2fe16c30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 14:57:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca903a733e48c17e82c792cddac0dcff5df75d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:10:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43a642aac66b480f921571932c35a547 : SUCCESS in 6m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4ff2fa7986c4973851be3fdc62e9b6b : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b682ca56b548442985167c644249c6f0 : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e68eecd3aa01455ca205a49c023d9f81 : SUCCESS in 11m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2fdc495f612e43b48843a92c36a8bf27 : SUCCESS in 11m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f1ffefa91dd4a4bbd3f53fa87c73a87 : SUCCESS in 10m 47s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3267a511da834339889820bc288eb00f : TIMED_OUT in 2h 02m 20s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e6b1c98711ea4fa89e53f1afd52f6d01 : SUCCESS in 23m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85d18371aacc4cf3b1003d287abd3efa : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7bb925c96a521890b956c83f6759fdb393a99ee3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-27 17:14:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"656450692f3ee55b9315fdb465110d0a2d96fe83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:28:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3de83147ea954097b6efce84d5f3e2af : SUCCESS in 4m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ade0a19404f04a99be115cb12df27d9e : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63cb5cb6434444b2a758f705c22d1cff : SUCCESS in 5m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8c8e4116e0ed45d89d9ebab4ca4be01f : SUCCESS in 12m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/917dfba45b4b4f3b8c43a6854b1ee728 : SUCCESS in 15m 11s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b42656fdc38b4140b49e05ca5cc64397 : SUCCESS in 11m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1a0d34a5ace439fbe327f7ac8e02889 : SUCCESS in 12m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6384c91970d347b7ab3ba423bfa34738 : SUCCESS in 46m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e779fadd5d3340aa8a069dfc8940bef4 : SUCCESS in 1h 07m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/32d969fb86ed4239b53ed895d5120c45 : SUCCESS in 24m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f202c18642fe4c02adfcf9d8898a355a : FAILURE in 12m 38s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8747f10cc1914ed68eb432cc8c22b57a : FAILURE in 10m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87f1fef8454b466db425dad890788811 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"19a0900ab41211bf949c418c267b756bd63638b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:28:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26bbef6ee27480f422450273f713ee0383566c33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:47:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"265cd8e611e6aae07b5f6660c3239009e3178090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:47:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b162101b041b4fbca1e1e1547115ef7f : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b8bcde463794702ab2eec549a8a0206 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1944ece90a9746fcb0cb59abb42065e4 : SUCCESS in 4m 56s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5c28862ec6a8478c87de3a5cab50a383 : SUCCESS in 13m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f2bf890045234f9295d66c690b99dd1a : SUCCESS in 12m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce5dacd71010456eb78e77d279eb2794 : SUCCESS in 13m 18s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3ca4d50c6dba4657906cc80a972a6bb7 : SUCCESS in 1h 15m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/12dc9b80d9b640eea6c5524af008a0aa : SUCCESS in 21m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ade02a733ac94ae4bd755e7901e5a90b : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3c2597063b16b739a915e6b8f66f5efa307858a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:49:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f3dbefe95834d729db3414b99b9875f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3134f475ab516887d25792764bd4cd7e3bff6d9f","revisions":{"6f3163a2f0415ac1aabf9162f9d4e5b8f03631b3":{"kind":"REWORK","_number":1,"created":"2020-04-25 14:29:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/723100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/723100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/1"}}},"commit":{"parents":[{"commit":"8b7dea0f048c75a753152e958ad165da12908dba","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8b7dea0f048c75a753152e958ad165da12908dba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-25 14:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-25 14:26:44.000000000","tz":540},"subject":"Deprecate designate::worker::worker_notify","message":"Deprecate designate::worker::worker_notify\n\n... because its actual parameter in designate was already deprecated in\ndesignate.[1]\n\n[1] 52ad1c4c13d1fead646435634e4b07bb32c68d28\n\nChange-Id: If22e4f75a0959507d6d3a97036f873df2f284d03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6f3163a2f0415ac1aabf9162f9d4e5b8f03631b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6f3163a2f0415ac1aabf9162f9d4e5b8f03631b3"}]},"branch":"refs/heads/master"},"3134f475ab516887d25792764bd4cd7e3bff6d9f":{"kind":"REWORK","_number":2,"created":"2020-04-25 14:49:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/723100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/723100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/723100/2"}}},"commit":{"parents":[{"commit":"8b7dea0f048c75a753152e958ad165da12908dba","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8b7dea0f048c75a753152e958ad165da12908dba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-25 14:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-25 14:49:28.000000000","tz":540},"subject":"Deprecate designate::worker::worker_notify","message":"Deprecate designate::worker::worker_notify\n\n... because its actual parameter in designate was already deprecated in\ndesignate.[1]\n\n[1] 52ad1c4c13d1fead646435634e4b07bb32c68d28\n\nChange-Id: If22e4f75a0959507d6d3a97036f873df2f284d03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3134f475ab516887d25792764bd4cd7e3bff6d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3134f475ab516887d25792764bd4cd7e3bff6d9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
