)]}'
{"id":"openstack%2Fpuppet-designate~725760","triplet_id":"openstack%2Fpuppet-designate~master~I121e06e5586d574aa63e4fd0b6553d414e227caf","project":"openstack/puppet-designate","branch":"master","topic":"v1-api","hashtags":[],"change_id":"I121e06e5586d574aa63e4fd0b6553d414e227caf","subject":"Remove v1 api support","status":"MERGED","created":"2020-05-06 05:29:32.000000000","updated":"2020-05-07 15:25:17.000000000","submitted":"2020-05-07 15:23:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725760-1588864999011-39316be2","meta_rev_id":"24f6d2154c739c2ef32eb0d21eb8d61053cb7a1d","_number":725760,"virtual_id_number":725760,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 15:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-07 14:17:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-06 09:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-05-06 22:35:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-07 14:17:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 09:04:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-06 22:35:19.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2020-05-07 14:17:29.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-07 15:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2561aab1cdb51a4e179afc82b97e48b345c7d91","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 05:29:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4d8325b7cb8136398f4c852d4b2fcfd8ff4e23f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 05:58:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fae2b3181d5a743f35f5f4b876ac2b33ebf87586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 07:05:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed646f5c310e44c48423bfdd2b28e352 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96219f63162c497c9b8e3fc5833ead73 : SUCCESS in 7m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8cff2450fbe45e3972313ae2ac2db09 : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/576521f3cf1a449eb120fbe3d4548cc2 : SUCCESS in 14m 11s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c1684395a89947a59d32ed2e79a4c995 : SUCCESS in 16m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/755f00451fb34e80934fe40f7ca7f85c : SUCCESS in 11m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efbd1ac8bb094ed0b9093889fadc0b28 : SUCCESS in 13m 41s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3925fd7a9d43488fb4403fdfe2dc3c1c : SUCCESS in 51m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1c9c914962244ba8b94730416c8b2c9e : SUCCESS in 1h 05m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/440b0107484941a18b073b29cdeb18ed : SUCCESS in 22m 04s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0d82fff7332e464d93d62b7782a34a8f : FAILURE in 13m 05s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/552902a2c57343878571e25d1758ce14 : FAILURE in 8m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/381c22b9df344752b06062321e8f7fb2 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"fac9ccbfa9ca1db2cc18c0acc26c6215edfca22f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-06 09:04:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1d1c4ee37e6ad56e0809686a7f8f5d924ccb0a74","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-06 22:35:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7b52e362e72ef480231586034baf96a0426cfc72","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-07 14:17:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18ab626b8528e86ddeaf2d16d5982e2371fd35d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:19:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d2b78ea9cdd4338f2ec4ef5f371f51e276aa544a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:23:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ec05d9da8944723a16bafefab287b7b : SUCCESS in 10m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af6e48054ac34f4d8f1280777830c324 : SUCCESS in 8m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b1fd8f050dc4a5a807de32d10c0c466 : SUCCESS in 5m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/acc094de03e04ec2a95c32beb890dcca : SUCCESS in 12m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a73042afe99341b6897fb56851b42805 : SUCCESS in 16m 19s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e4ed04be70147679ccf4c305d9aea35 : SUCCESS in 13m 02s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/75381cdc21d14a639537cb52bbfa56b5 : SUCCESS in 1h 01m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1d49a156172d4329adeb82599c0237e2 : SUCCESS in 20m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ddd7ce1e9754b6ca15bb01769feffd5 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b76c2d67273e51761c9a14a39766a852fad5ccc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:23:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"47091f3b59456d46e8c19452ac19a76ff8c575b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:25:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/759abc2f5d14481f9da20b79f152aefc : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c794aa6e64f13840db6f8870e22d51e21be81e58","revisions":{"18e55af3d08535bd0c3c60af2aa61beda41a4021":{"kind":"REWORK","_number":1,"created":"2020-05-06 05:29:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/725760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/60/725760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/1"}}},"commit":{"parents":[{"commit":"3f01e55290d231657b6c6aa37761e582a7687f13","subject":"Runs on Debian 10.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3f01e55290d231657b6c6aa37761e582a7687f13"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 05:28:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 05:28:42.000000000","tz":540},"subject":"Remove v1 api support","message":"Remove v1 api support\n\n... becase it was deprecated a long ago in puppet-designate, and actual\nimplementation was already removed fron designate[1].\n\n[1] c318106c01b2b3976049f2c3ba0c8502a874242b\n\nChange-Id: I121e06e5586d574aa63e4fd0b6553d414e227caf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/18e55af3d08535bd0c3c60af2aa61beda41a4021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/18e55af3d08535bd0c3c60af2aa61beda41a4021"}]},"branch":"refs/heads/master"},"c794aa6e64f13840db6f8870e22d51e21be81e58":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-06 05:58:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/725760/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/60/725760/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/60/725760/2"}}},"commit":{"parents":[{"commit":"3f01e55290d231657b6c6aa37761e582a7687f13","subject":"Runs on Debian 10.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3f01e55290d231657b6c6aa37761e582a7687f13"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 05:28:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 05:58:27.000000000","tz":540},"subject":"Remove v1 api support","message":"Remove v1 api support\n\n... because it was deprecated long ago in puppet-designate, and actual\nimplementation was already removed from designate[1].\n\n[1] c318106c01b2b3976049f2c3ba0c8502a874242b\n\nChange-Id: I121e06e5586d574aa63e4fd0b6553d414e227caf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c794aa6e64f13840db6f8870e22d51e21be81e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c794aa6e64f13840db6f8870e22d51e21be81e58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
