)]}'
{"id":"openstack%2Fpuppet-designate~725762","triplet_id":"openstack%2Fpuppet-designate~master~Idb4187b7d23b62a5c7aa575c2f4501e8e4a653a4","project":"openstack/puppet-designate","branch":"master","topic":"warning-typo","hashtags":[],"change_id":"Idb4187b7d23b62a5c7aa575c2f4501e8e4a653a4","subject":"Fix typo in warning message","status":"MERGED","created":"2020-05-06 05:34:11.000000000","updated":"2020-05-06 11:58:32.000000000","submitted":"2020-05-06 11:58:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725762-1588766312493-fb73416c","meta_rev_id":"7b6b65684024c887f76f5e6e5e7fb6ab744cf1a3","_number":725762,"virtual_id_number":725762,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 11:58:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-06 10:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-06 09:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-05-06 10:48:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 09:04:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-06 11:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58798b8e963896f2bd59a59bfb221be44a7b9126","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 05:34:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d64f02142461e2dd42576776aadcd48821c8dcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 06:44:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c733247543e40fba3a3a5fa00029edb : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f440ebbfe2dd40d0853bd1df39e97c31 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40825a8bfad94ccf863c83aedfed54d7 : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1226d7ca39764eaba84d05005624d15b : SUCCESS in 11m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/aca48757b4e843189cb5c7d13abd2b9e : SUCCESS in 10m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a711083b1cd4f1696b6f86d57c137c2 : SUCCESS in 9m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd16c055caa94f70beb3ca114f11289a : SUCCESS in 8m 27s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a81870edf44a4af1b6aa0db44ce918aa : SUCCESS in 42m 41s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8978285dfb1c477d9fc38271613ebf72 : SUCCESS in 1h 06m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/739a0eba48834c6aa6990d476494080e : SUCCESS in 17m 42s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/477e1f4166c84097b3804739315fa53f : FAILURE in 8m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6c6275f809c43188c8b0333991ac208 : FAILURE in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe2eccce004452dfa980f79d39138e5804155ece","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-06 09:04:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e4d765b3399fe386e4c946906b3be2c4e70a9e1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 10:48:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is a trivial fix to correct a typo, so let me approve this by myself.","accounts_in_message":[],"_revision_number":1},{"id":"2d074b1339b1c1f751bcb65cdf2fb1a56cdf6e4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 10:48:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7eaa5390ffc69d1852ea4833b090cd78caf06cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 11:58:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ad6a49a591330a1993b04aca5225ec781debf0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 11:58:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d6c038d3242469e9f8165f63196ce9a : SUCCESS in 5m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea3e2f7f60274d24a8bc2f81b687be65 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c738d8605554dc0ab2d22ef7b7d211d : SUCCESS in 6m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/def0b1b99bbe42babe61f55e73a5adb8 : SUCCESS in 11m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/59a6b1c3d249439093781878aee69574 : SUCCESS in 12m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51e48ce26f5b44c0b26be81e3679990e : SUCCESS in 10m 31s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c3b4976db5a4031b31b3df2f4b9d7ce : SUCCESS in 1h 05m 41s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/af816b826e46474fad127f7f180187ab : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d997168369fee531ba95714f251fc3ceff37901e","revisions":{"d997168369fee531ba95714f251fc3ceff37901e":{"kind":"REWORK","_number":1,"created":"2020-05-06 05:34:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/725762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/62/725762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/62/725762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/62/725762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/62/725762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/62/725762/1"}}},"commit":{"parents":[{"commit":"3f01e55290d231657b6c6aa37761e582a7687f13","subject":"Runs on Debian 10.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3f01e55290d231657b6c6aa37761e582a7687f13"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 05:33:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 05:33:53.000000000","tz":540},"subject":"Fix typo in warning message","message":"Fix typo in warning message\n\nTrivial-Fix\n\nChange-Id: Idb4187b7d23b62a5c7aa575c2f4501e8e4a653a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d997168369fee531ba95714f251fc3ceff37901e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d997168369fee531ba95714f251fc3ceff37901e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
