)]}'
{"id":"openstack%2Fpuppet-designate~728173","triplet_id":"openstack%2Fpuppet-designate~stable%2Fussuri~I8405576e444b77941e738616c6ccafe9305f036b","project":"openstack/puppet-designate","branch":"stable/ussuri","hashtags":[],"change_id":"I8405576e444b77941e738616c6ccafe9305f036b","subject":"Switch to Ussuri","status":"MERGED","created":"2020-05-14 17:56:22.000000000","updated":"2020-05-18 07:30:33.000000000","submitted":"2020-05-18 07:30:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728173-1589787033081-0ef704bd","meta_rev_id":"eb11cb0096c59ef346eac0c2b9891a03587b3edd","_number":728173,"virtual_id_number":728173,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 07:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-05-18 06:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-05-15 22:48:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-05-18 06:01:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:48:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-18 06:01:31.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-18 07:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9c590ca1fdcd517fb40484a8e680bd0de851247","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 17:56:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a656a651fb3d0a7d8bd9b63ffee1f1a8c8ed556e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:06:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49c92dcafecc4d51984529b3e7dd872d : SUCCESS in 6m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f406781603424aee901f29ffa3bb1dec : SUCCESS in 6m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b1dd4eb79d646c192193ede3c9c3c86 : SUCCESS in 6m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/272d92251ceb4a6588a31525e57eec0a : SUCCESS in 15m 56s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/20af44267b494dbb851d986802b0cc63 : SUCCESS in 16m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4136254444434c289e810ee208041942 : SUCCESS in 15m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07040b8f6d8147fba6670cdd8693c02a : SUCCESS in 14m 48s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78f2bc00b417440f911abddd0a84610a : SUCCESS in 55m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/db8c8d482ce045f49e2665e7bf6cc3ce : SUCCESS in 1h 04m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1e78046bfb57431ab33acb3f79275bd1 : SUCCESS in 24m 10s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b56fd7be0b6c4f84803ba1cce9ffbc38 : FAILURE in 12m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d1a2ca1de4b49428104cb54e6c740ea : FAILURE in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75b9ba7e8ecb0bc32e01883b3102878c44c1a8bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 22:48:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"718e0e894bbbffa1f749eecbc31a512ef06287ca","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-18 06:01:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"811c32eadcb9043d5b5f70ad3df4459001b89775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:01:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c80ef5bdd85b8e09639724d72882d8df8abb535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 07:30:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c925e9d404ad40eb9ae9b02273ff7a04 : SUCCESS in 5m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35e8b6dde3c1428ba32caf685f931506 : SUCCESS in 7m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4d636e1311c43aa80d02e6b717822bd : SUCCESS in 7m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/28c0d209a7bf4980af04ff8d7f9e3144 : SUCCESS in 16m 04s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/11204b25ebf14946ac31207fc691df4b : SUCCESS in 14m 46s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a66c73fd33fa4cd0a0beefcc165f9e6f : SUCCESS in 15m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eb3e0231e02c40a18bd5401473eb0e35 : SUCCESS in 1h 19m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/85cd2180aad241d094a68f96518910e5 : SUCCESS in 23m 11s","accounts_in_message":[],"_revision_number":1},{"id":"eb11cb0096c59ef346eac0c2b9891a03587b3edd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 07:30:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16b4663a2fa3416d3033fef3ae0910de5c1644f6","revisions":{"16b4663a2fa3416d3033fef3ae0910de5c1644f6":{"kind":"REWORK","_number":1,"created":"2020-05-14 17:56:22.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/73/728173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/73/728173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/73/728173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/73/728173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/73/728173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/73/728173/1"}}},"commit":{"parents":[{"commit":"944f5f4cb9a0dbfbf8fa63015a42b069f6bec3f2","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/944f5f4cb9a0dbfbf8fa63015a42b069f6bec3f2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 17:56:12.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 17:56:12.000000000","tz":120},"subject":"Switch to Ussuri","message":"Switch to Ussuri\n\nChange-Id: I8405576e444b77941e738616c6ccafe9305f036b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/16b4663a2fa3416d3033fef3ae0910de5c1644f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/16b4663a2fa3416d3033fef3ae0910de5c1644f6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
