)]}'
{"id":"openstack%2Fpuppet-designate~745473","triplet_id":"openstack%2Fpuppet-designate~master~Idb5ecfd027728dc0c85ffac5ce08d475e6ea8cb6","project":"openstack/puppet-designate","branch":"master","hashtags":[],"change_id":"Idb5ecfd027728dc0c85ffac5ce08d475e6ea8cb6","subject":"Add backend coordination url option to producer","status":"MERGED","created":"2020-08-10 03:34:06.000000000","updated":"2020-08-10 18:25:46.000000000","submitted":"2020-08-10 18:25:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745473-1597083946193-12991b16","meta_rev_id":"c5719cd24a58e19f83a676f31a315b3785f0d8aa","_number":745473,"virtual_id_number":745473,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-08-10 13:58:14.000000000","_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"date":"2020-08-10 13:34:33.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 18:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-10 12:50:26.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-08-10 17:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-08-10 17:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 12:50:26.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-10 13:34:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-10 17:00:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-10 18:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a433b5bd4d2a424d821305cca01ba56f7c6d90c","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-08-10 03:34:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"963e7773fef8ef265f869c413cc2bad3d041edb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 05:10:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e32890d492f046c5b4d73f3d0f48d94d : SUCCESS in 3m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06d3f475df4d423189beb5c42e489d94 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2838d667194e41bbbc908c798211ed3f : SUCCESS in 5m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5c858ed90589499294339a9fbf85a2d1 : SUCCESS in 15m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/daf26b7a357b4c32956c57aaa463c909 : SUCCESS in 12m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62eb231ac1cc46d4978ec61a0edde25f : SUCCESS in 8m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b39bc782d8b34fc08692dbbbc837f0a1 : FAILURE in 7m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fd0fefc834d42fdbdc443de20c6cd40 : SUCCESS in 42m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/687482671bd2474e9c16753325608c22 : SUCCESS in 1h 29m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e6f873dd252e4d859399336a0a777d2f : SUCCESS in 25m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6bb6611f618d451bb22f86659cd0bb40 : FAILURE in 7m 43s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da5cafa4560c4107a141c888ce1ed538 : FAILURE in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2bfa7b8b7b6c5a2ff44f7f30a439b79ca1a8add","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 12:50:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0111191a5fb41fcd6928724d2742b5fd49fd1d9b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-10 13:34:33.000000000","message":"Patch Set 1:\n\nIs this really specific to the \"producer\" or shouldn\u0027t the [coordination] section be more generic?","accounts_in_message":[],"_revision_number":1},{"id":"aaccef6882a5671ccae760b7bd29c841f767a47e","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-08-10 13:58:14.000000000","message":"Patch Set 1:\n\n@Tobias I think the producer is the only one that uses coordination but I\u0027ll double check","accounts_in_message":[],"_revision_number":1},{"id":"4906d5ca64bcabdc9084083daac915341b7ade4a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-10 17:00:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62a351184cf63e3c5e60994a4e475114010287ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:01:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb0e57b508cd32f48d02e9b745ffb39058de94e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 18:25:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a02dbc673a1e44ed99cdf2e549e0d14e : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cff927fca4bf4c3ebb73534fd6aa75c5 : SUCCESS in 7m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5e50a09f9a140aeb307978da3b5ff03 : SUCCESS in 4m 04s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cdd07423bc024b0c8f4ce5ce1bf99b69 : SUCCESS in 10m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/60ae6e013e7e4293b679f493f35dfb90 : SUCCESS in 11m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81978b41a91d4822804adeae23fbbd23 : SUCCESS in 13m 05s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/790ee124417a483fb55cf461e316b292 : SUCCESS in 1h 18m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c0a854b3dda0420b967db490f6b7c2b3 : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c5719cd24a58e19f83a676f31a315b3785f0d8aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 18:25:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba9fa2ad2e31a969b4c481d7459d0fdc4c6adece","revisions":{"ba9fa2ad2e31a969b4c481d7459d0fdc4c6adece":{"kind":"REWORK","_number":1,"created":"2020-08-10 03:34:06.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/73/745473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/73/745473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/73/745473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/73/745473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/73/745473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/73/745473/1"}}},"commit":{"parents":[{"commit":"b97e1961c98431ec9a453736cb56969898884444","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/b97e1961c98431ec9a453736cb56969898884444"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-08-10 03:31:05.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-08-10 03:33:59.000000000","tz":600},"subject":"Add backend coordination url option to producer","message":"Add backend coordination url option to producer\n\nProducers that are deployed in an HA environment that do not have\na coordination url set will produce multiple copies of the same\nthing and operate in standalone mode.\n\nChange-Id: Idb5ecfd027728dc0c85ffac5ce08d475e6ea8cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/ba9fa2ad2e31a969b4c481d7459d0fdc4c6adece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/ba9fa2ad2e31a969b4c481d7459d0fdc4c6adece"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
