)]}'
{"id":"openstack%2Fpuppet-designate~749105","triplet_id":"openstack%2Fpuppet-designate~master~I8a5b608735cffe1fc7ca5fcd51930cfbe8b3e0eb","project":"openstack/puppet-designate","branch":"master","hashtags":[],"change_id":"I8a5b608735cffe1fc7ca5fcd51930cfbe8b3e0eb","subject":"Add Puppet Litmus","status":"MERGED","created":"2020-08-31 19:16:35.000000000","updated":"2020-09-10 01:31:19.000000000","submitted":"2020-09-10 01:31:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749105-1599701479950-a980bdbc","meta_rev_id":"33ecb1e23595d4763557040c0b93305cc058905c","_number":749105,"virtual_id_number":749105,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-09-09 22:55:31.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 01:31:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 18:27:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-08 18:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 18:27:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-09 21:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 18:27:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-09 21:28:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-10 01:31:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"249b0f1bcddd25c2c8f1a118939179c3935ffa48","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 19:16:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2ea430f13a2f352033109d26304f35a1e578c3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 20:50:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f353a7ff6ca74d0d891c0ea63f1a33aa : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a11f1306d0734580937581fdc8a49747 : SUCCESS in 3m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e4459cdd75b34162b3d51441519bd0b4 : SUCCESS in 9m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecb0c805327845b48561fd8d051668c0 : SUCCESS in 8m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8311c58ed7744c419700306e8a76233a : SUCCESS in 9m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2177339c5e54bb09a92f07e2c354cda : SUCCESS in 43m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/045c63499c7a460a9138e1e8f8b95c65 : SUCCESS in 1h 28m 53s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e584975976b04eb0aebbb8dfc46fa466 : SUCCESS in 18m 00s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/512eda64991e47c1b55fe21141ffb97f : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2e022cc77052dbef035fd87bd01ab1bc453ad49b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:18:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4c3252f3ee4846ae213cb6858185cdf7e371bb2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:27:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bafe77ba11fe02f4f9043cd5df326cdf57e33224","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-09 21:28:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfbbc33490e6754693991380eb77a8e96c2d4b76","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-09 22:55:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e32731fb6dd805eabca822eb021442fa96d08a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:15:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9be9ca5565b94a77a61ec5d733c7ddf4 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d0e26f5505946f787a4f666ffcb8398 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/897ce0bba4274f62b2564afb18814a70 : SUCCESS in 11m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96d8cc8e47b14ba7b039ab2bdb55b768 : SUCCESS in 7m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67422f056cc24af7a6918dac5a7354fd : SUCCESS in 9m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb0356b3bf944d64ac8d6ebfa101d39e : SUCCESS in 45m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ba159165aff342a29291973f33565c86 : SUCCESS in 55m 07s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e32bb1d5b30d4587a87abe96c85fff66 : SUCCESS in 19m 31s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4e592892fb248fd969b4ac586ac23be : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"401101a7cc9092b037e98d51bf4f8115cabd3310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:15:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21a4b65e4fe3aead8070861f3b2ded39f4576eea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:31:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"33ecb1e23595d4763557040c0b93305cc058905c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:31:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9b82364ed23404ca5da54065b3c554a : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07d55ca2c25843d4951902670c662841 : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3085468a035840eda77c1c6a727e14fc : SUCCESS in 10m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd6aa06c32f845dc8fb738b4c3aae844 : SUCCESS in 8m 03s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5f45610c90cd4d7189db9d952a34b6b9 : SUCCESS in 1h 07m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/10a358db61474c83b710e38da9077b11 : SUCCESS in 17m 29s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1931e3fc678f4a8cafc1e724a0c6b564 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40700887a6d5cc20b61a5927b376ee04bf36192d","revisions":{"40700887a6d5cc20b61a5927b376ee04bf36192d":{"kind":"REWORK","_number":1,"created":"2020-08-31 19:16:35.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/05/749105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/05/749105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/05/749105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/05/749105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/05/749105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/05/749105/1"}}},"commit":{"parents":[{"commit":"319615572ba76615d0acde5e0aefb031cf614a3f","subject":"Merge \"Use oslo::db class to set database parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/319615572ba76615d0acde5e0aefb031cf614a3f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:16:26.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:16:26.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: I8a5b608735cffe1fc7ca5fcd51930cfbe8b3e0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/40700887a6d5cc20b61a5927b376ee04bf36192d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/40700887a6d5cc20b61a5927b376ee04bf36192d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
