)]}'
{"id":"openstack%2Fpuppet-designate~763694","triplet_id":"openstack%2Fpuppet-designate~master~I0479e0ef6262d6753d199c12fb0dc98069dd32fe","project":"openstack/puppet-designate","branch":"master","topic":"bug/1904962","hashtags":[],"change_id":"I0479e0ef6262d6753d199c12fb0dc98069dd32fe","subject":"Allow db sync timeouts to be configurable","status":"MERGED","created":"2020-11-23 01:39:41.000000000","updated":"2020-11-25 12:51:05.000000000","submitted":"2020-11-25 12:49:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763694-bug/1904962","meta_rev_id":"dd9d2677ac73ba40728d4fee812af314eef24c98","_number":763694,"virtual_id_number":763694,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 12:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-11-25 11:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-11-24 15:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-11-25 11:13:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 03:12:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 15:16:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-25 10:54:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"cb5c6fc881ca931fe7cd9e488476f646d6fc68b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-23 01:39:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"772f8c3e9e8f66a757c8ee7180af321312e10de9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 03:12:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/192147d6e8e24106a30069e30cb1d62b : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8fccb637fc243b785d4cb0c1dfbc710 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/786103c4b034408b817216f69ce08233 : FAILURE in 8m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b917249f74d4cf993b3729dd8c420ea : FAILURE in 6m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c095203e63654a30800ea63f9a5c1157 : FAILURE in 5m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae161553561b4bec807bf4580eec8a7d : SUCCESS in 48m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e9d5a73ae02748238c679206b42b0381 : SUCCESS in 1h 25m 47s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/cbc3c1333b0e4a78b1c938489a2e8923 : SUCCESS in 18m 21s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f42d63c001344e8bb73bfb9f7b1d46bb : FAILURE in 15m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c3e8246978f41dc89270511509dbc29 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2eaf8fe9937e9edec9dfe1539e00bc7f927c4c6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-23 05:30:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19e37e437d886d429d7a1114cebfa13b0ae8b4cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 06:53:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35901e307aa4484a87767816ba05a106 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dc74ade65df408085cf95ee79306c53 : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/184061dd2dc94a5c9fb467e04cd08e87 : SUCCESS in 10m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a50b650681d340a3b63709d09306f891 : SUCCESS in 9m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a324b8cad1284ea78ebe9c9dc5632096 : SUCCESS in 9m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b9f555a7b6e4fd484dd6404721b9fcd : SUCCESS in 50m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/14d722fbfe9b4bf6849dfb6fc03324b6 : SUCCESS in 1h 20m 25s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4f79c31ceffb4168a410d3c53ff2aaad : SUCCESS in 22m 01s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50f4921d6e8f4a1ca8c566e762821b76 : SUCCESS in 15m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48d670b238584f2c80903bf755784e9c : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f435802c03c2893d262273a1db7ecee84ae6f9af","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 15:16:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6f019a44ff2ab0e229aaf97c6ea7dfccaa3ce26d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-25 10:54:37.000000000","message":"Patch Set 2: Code-Review-1\n\nAdding -1 since my comment for another patch[1] is applied to this patch as well.\n\n[1] https://review.opendev.org/c/openstack/puppet-openstack-cookiecutter/+/763698","accounts_in_message":[],"_revision_number":2},{"id":"f87a0097512344e0410a2748c3ef0bc51e6356f1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-25 11:13:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0d8181284da7e6e90ced847d4bf951b7595bdb3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:16:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4de17424bfd9fa74392a38ce5787c607b0cb4f33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:49:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9991231278b4898ad20738d56a70f11 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1d2436a31bf4f578328e2df579b840d : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/99389e18cbca45cca110b421d591da5b : SUCCESS in 12m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05195d219ff14506bb20764562990319 : SUCCESS in 10m 07s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e9f193228b284e47951ac967aa4388c2 : SUCCESS in 1h 25m 30s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5976ad0e36434a1aa52f2cefc89beb2a : SUCCESS in 21m 04s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1271581a5ba94d4fb784c4d74f489aa9 : SUCCESS in 16m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/095629946d1d42f18b3fa3248ea7a503 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e5af7e72298df04b3bafa37c4306c90411e721db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:49:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dd9d2677ac73ba40728d4fee812af314eef24c98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:51:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bceb8e087b15406db534e14df5162612 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"405eac4390ecf6780e55b08a5a39c6d96408a026","revisions":{"f16da7cf7343742535cec96c31d15168e1c0bc6b":{"kind":"REWORK","_number":1,"created":"2020-11-23 01:39:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/94/763694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/94/763694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/1"}}},"commit":{"parents":[{"commit":"982102a71878e35d55ee0b8d2a7d55effcef7353","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/982102a71878e35d55ee0b8d2a7d55effcef7353"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"subject":"Allow db sync timeouts to be configurable","message":"Allow db sync timeouts to be configurable\n\nAs Openstack projects continue to have longer database migration\nchains, the Puppet default timeout of 300 seconds for an execution\nis becoming too short a duration on some hardware, leading to timeouts.\nAs projects continue to add more migration scripts without pruning\nthe base, timeouts will continue to become more frequent unless\nthis time can be expanded.\n\nChange-Id: I0479e0ef6262d6753d199c12fb0dc98069dd32fe\nCloses-Bug: #1904962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f16da7cf7343742535cec96c31d15168e1c0bc6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f16da7cf7343742535cec96c31d15168e1c0bc6b"}]},"branch":"refs/heads/master"},"405eac4390ecf6780e55b08a5a39c6d96408a026":{"kind":"REWORK","_number":2,"created":"2020-11-23 05:30:07.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/94/763694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/94/763694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/94/763694/2"}}},"commit":{"parents":[{"commit":"982102a71878e35d55ee0b8d2a7d55effcef7353","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/982102a71878e35d55ee0b8d2a7d55effcef7353"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 05:17:14.000000000","tz":480},"subject":"Allow db sync timeouts to be configurable","message":"Allow db sync timeouts to be configurable\n\nAs Openstack projects continue to have longer database migration\nchains, the Puppet default timeout of 300 seconds for an execution\nis becoming too short a duration on some hardware, leading to timeouts.\nAs projects continue to add more migration scripts without pruning\nthe base, timeouts will continue to become more frequent unless\nthis time can be expanded.\n\nChange-Id: I0479e0ef6262d6753d199c12fb0dc98069dd32fe\nCloses-Bug: #1904962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/405eac4390ecf6780e55b08a5a39c6d96408a026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/405eac4390ecf6780e55b08a5a39c6d96408a026"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
