)]}'
{"id":"openstack%2Fpuppet-designate~773457","triplet_id":"openstack%2Fpuppet-designate~stable%2Ftrain~I0fc8fbff2e723434f400d5fa4721ec6092293b7c","project":"openstack/puppet-designate","branch":"stable/train","topic":"set-sqlite-db","hashtags":[],"change_id":"I0fc8fbff2e723434f400d5fa4721ec6092293b7c","subject":"Add support to specify sqlite database","status":"MERGED","created":"2021-02-01 17:34:02.000000000","updated":"2021-02-03 11:09:27.000000000","submitted":"2021-02-03 11:09:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773457-set-sqlite-db","meta_rev_id":"1b971f304191f8bc9b75d2b7829d8e2a78cc88ef","_number":773457,"virtual_id_number":773457,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 11:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-02-03 10:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-02-03 10:15:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 18:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 08:32:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-02-02 08:32:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"},{"updated":"2021-02-03 10:15:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"dbed786081825117d5f37395371bf1b57f1cb30c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-01 17:34:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d940df8a1563b39664e10c2ecf7c52b236d5e1a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-01 17:35:10.000000000","message":"Topic set to set-sqlite-db","accounts_in_message":[],"_revision_number":1},{"id":"c960a346507c3b95b9e0f9d705dd6cd6e15eac0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 18:40:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31c708f8048f4bb49053092a8ac268a5 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54e7a5bd5e1d49058ae635e646ede193 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8aea1dd4c7824d8ea11dcde672db0379 : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/440500c51b8a4f29aa01f827fe507b69 : SUCCESS in 12m 42s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/422065c14a1c430a8daab358b490a6dc : SUCCESS in 8m 44s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/fe1ccfd795e54f5c8c5d1ad50c954cae : SUCCESS in 9m 21s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fc44d979ec8420f83f61925cc51195d : SUCCESS in 7m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55306aea3a9e487c8549defb33a53565 : SUCCESS in 9m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/729e3103f5284d7481783da72eebb06b : FAILURE in 37m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ad92eb7c6ada4f99b452b98033e463aa : SUCCESS in 57m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/26db7bb0bb31408ca2bef16a4b109a99 : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/06096f96ac724c7cb167f6c9006a0010 : SUCCESS in 16m 01s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b49853601f164397996a1e9b555dfb5c : FAILURE in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"623c1b777a22f7fa73287b7837d93ccdeba7f8ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-03 10:15:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThe change is minimum and the backport to ussuri has been merged.\nSo let me approve the change by a single core.","accounts_in_message":[],"_revision_number":1},{"id":"f15d4a417219cdc3ba7724885cd5607249014071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 10:16:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70a916fb53ce3f6107bf2fe103bd3bcd09b2ccb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:09:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45d4143750d34396abe8a16f69ca44c6 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56efdf3ee438474d95363477024e7337 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d75a55ee9574a24bae149afaaece2d3 : SUCCESS in 4m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ffc10eadb50741cab3ea75e3c9623d68 : SUCCESS in 16m 54s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f261d43731c43e08aa67d78d426b924 : SUCCESS in 7m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/126d0e6376634b0fba2f3a12a663e8a6 : SUCCESS in 50m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/18b17b22922245dfb2b42b802d14c92d : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1b971f304191f8bc9b75d2b7829d8e2a78cc88ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:09:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e554a108f56b5403d813722a3fc68d9464aa3f2","revisions":{"3e554a108f56b5403d813722a3fc68d9464aa3f2":{"kind":"REWORK","_number":1,"created":"2021-02-01 17:34:02.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/57/773457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/57/773457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/57/773457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/57/773457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/57/773457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/57/773457/1"}}},"commit":{"parents":[{"commit":"0543a1c2a0108fb7b67c41e6bc2fbad99eb24cb9","subject":"Release 15.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0543a1c2a0108fb7b67c41e6bc2fbad99eb24cb9"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-02-01 17:33:13.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-02-01 17:33:56.000000000","tz":60},"subject":"Add support to specify sqlite database","message":"Add support to specify sqlite database\n\nNeeded when running testing when you don\u0027t want\nto set MySQL bindings etc and just set the option.\n\nChange-Id: I0fc8fbff2e723434f400d5fa4721ec6092293b7c\n(cherry picked from commit 318ace33f79a79b6034271610485b3b6a037011a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3e554a108f56b5403d813722a3fc68d9464aa3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3e554a108f56b5403d813722a3fc68d9464aa3f2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
