)]}'
{"id":"openstack%2Fpuppet-designate~781756","triplet_id":"openstack%2Fpuppet-designate~stable%2Fussuri~Ief63f88ba323e3f20300b8ff6a5148a78413ac3d","project":"openstack/puppet-designate","branch":"stable/ussuri","topic":"postgresql-ussuri","hashtags":[],"change_id":"Ief63f88ba323e3f20300b8ff6a5148a78413ac3d","subject":"Add designate::db::postgresql class.","status":"MERGED","created":"2021-03-19 14:55:53.000000000","updated":"2021-04-01 17:57:04.000000000","submitted":"2021-04-01 17:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"781756-postgresql-ussuri","meta_rev_id":"2a88a0be9c2cfed29d920180204158c2dd592e8a","_number":781756,"virtual_id_number":781756,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 17:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-03-30 13:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-03-30 14:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-03-30 14:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 15:03:42.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-19 15:03:42.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-19 15:03:42.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-03-19 16:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8ff5e48eef8d86ebce095dd14be752d5f70b8f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-19 14:55:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"144d8ce7298cc9c683247a0f9a3c4ce93a6e1d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:06:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef444388d2ed4ca1bd6ce2e26c672b4f : SUCCESS in 3m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/febc46b047ee4e0d8aca8415a4e98351 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b51a544ca5e495289402324654f6c7c : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2c31d213f36b44baa50e22484d4509f7 : SUCCESS in 10m 51s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/45031632fb154a789e61579670fe45ba : SUCCESS in 7m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/614af40789ae47e78f1cc5d2bf0aae20 : SUCCESS in 8m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0eb70b49aca84268beb9e6aca8922d20 : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf4e9f9fa04e478ba0e64dcfa0def9cb : FAILURE in 37m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9552909064cf4218980bccccd2f36364 : SUCCESS in 1h 04m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6a518dcdd9d74323927f658e144e3db7 : SUCCESS in 17m 18s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a138f80eb154b00a0bd4decfa66eba2 : SUCCESS in 13m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10ae356da5bb4565972bb1d181cd6bf0 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3f78dea9bd32c7b50ff5dd71229839106343b4b2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-19 17:23:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"41fabb0115e48a1f899bfc9d587ecd8a14570b3f","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-30 09:49:55.000000000","message":"Patch Set 1:\n\n@Alex: Thanks for your +2. \n\nCould I be assigning this to any other reviewer? Apart from you Takashi or Tobias?\nI simply chose you because you did the change to master and stable/victoria","accounts_in_message":[],"_revision_number":1},{"id":"56154ca8449a9135605639a4b54e105dccb81709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-30 13:14:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e47c05906da871434b9baf15a6072b3fc4b1f8b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-30 13:17:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33bee3e39bed590a086da7e5677db581a8287bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 14:27:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05b7d24f232b4238aeac52249d5cd6c7 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4785312434a4f7a82b15972a2044140 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55fc69aeac5249cea28af09334b8fa81 : SUCCESS in 4m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c15de1cd5ad54a46ad47da1d0dd5304e : SUCCESS in 10m 18s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/cb323f39ee87450eb17225a0bfdcd41f : SUCCESS in 9m 49s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c77912a898ad4a6eb8283689a8435dbc : SUCCESS in 7m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/668b8b3314cd4c65b951bec1a27557bb : SUCCESS in 9m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9d8e5c28c2d4c67b652d286346208a7 : FAILURE in 31m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/16e3a0d6d73c4b36b8f88753f7f5828b : SUCCESS in 1h 05m 44s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ccfee54751204563af34df84f67be8bf : SUCCESS in 21m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88eb4832b289400d86705e1f84b30591 : SUCCESS in 16m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/323b0e1a4d5444a6a0214e028bc5ca0a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"685cdb5b522da412842da3944263356c101d3e3b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-30 14:29:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90d2689ba3c4ccbe39627fdd27297e3136bf1045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 14:31:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a279ca5b34f02e809224e0e8af4a0e3cebe800ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 16:44:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bb0f0d319fb4a769edbbe8a37d84dc7 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97176e2c68534d75bd2111badeaa3fd5 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3d36502729a4a3a8bfc5174966f047d : SUCCESS in 4m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b2a2e84066ee4559b49c9ebddba758a3 : SUCCESS in 10m 09s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a16f2910e33d45faa99968651dc0f33f : SUCCESS in 12m 16s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f26af359b7414e91ae59f1bd8a1c1e71 : SUCCESS in 6m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1cb47ce047e74170903a02f002fe4472 : RETRY_LIMIT in 1h 04m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6a62fdbf3f1e41d88dc0749c0dd7f26a : SUCCESS in 23m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71895ad520da48888980c507dc8d764b : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8bfb47215221b7c28be748f46f75a1e9e30324ef","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-30 21:34:22.000000000","message":"Patch Set 2:\n\nretest","accounts_in_message":[],"_revision_number":2},{"id":"9d7d8288bc53ee6e1e75c33e0e1d02b66b7224ca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-01 13:52:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"48f69b5fa03295e428d04963aa951d2a2e9787e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 16:46:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/492c42690e5a4bc9a7d7b7d316960d85 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/645ab779d5b943e08ffa25f6e7dc8c68 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5851f001443749b4b63e8f75abd81100 : SUCCESS in 3m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9a8c374735864b9eba2927f7b3279f5d : SUCCESS in 9m 49s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/6ac0371756cf4434b4f5947601fcb06f : SUCCESS in 9m 59s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0bcc12ff2f145ffb0ee7ce76e43c840 : SUCCESS in 9m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13be2a0298d046acbb0fc99de940e5d4 : SUCCESS in 9m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1cad8630ee247ad916cda40983fe50c : FAILURE in 35m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f1aa00a1274c4d8a8574c79fca1f17c7 : SUCCESS in 1h 03m 01s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c99bea2fbdf34149828b29a33faf6df6 : SUCCESS in 18m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3b9e6ca805046409f20a8ffe6c1bb05 : SUCCESS in 16m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39e6e740f6df4f10876f95fa1b666937 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"15d70f1b22028c33e26345b479a820d642ba9319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 16:47:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f173b99160ced56a459ad7b34f69d6a130fff02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:55:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53c1e99d6e864dda8527dcfed5804cba : SUCCESS in 3m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67b1c8c2585e4ec298a7ac7c3fae87bc : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d430051102c4094b9aa06c288e0d4ff : SUCCESS in 3m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f149a6dac8d24c4faa5d7a85e6a55e2a : SUCCESS in 8m 12s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c9a13ac94ef049ef914d243777b1292c : SUCCESS in 9m 24s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5edfd356068d43828c87d59dd0d3e6f1 : SUCCESS in 9m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f29f7bf5a3db46c7bb3f83cc7b13ea59 : SUCCESS in 1h 02m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7c3fee41755245faa2811e013cda22a6 : SUCCESS in 20m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9056fe0ed7554b61a761e7396065d855 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4e96b88fe86e015a86e4696291619a4f5b16f880","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:55:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2a88a0be9c2cfed29d920180204158c2dd592e8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:57:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e39c504e8574ac0905bf5e7dd73322e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"71b25d083e97caa32f0ab6c7adef4fb9344040c1","revisions":{"95799c3734bf0287ca809af12c221a4807451401":{"kind":"REWORK","_number":1,"created":"2021-03-19 14:55:53.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/781756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/56/781756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/1"}}},"commit":{"parents":[{"commit":"6f32bad5d0876a4661e644198a54a625f39942af","subject":"[Ussuri and before] Add support to specify PostgreSQL database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6f32bad5d0876a4661e644198a54a625f39942af"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-17 10:34:27.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-19 14:53:10.000000000","tz":60},"subject":"Add designate::db::postgresql class.","message":"Add designate::db::postgresql class.\n\nThis allows Designate to be used with PostgreSQL.\n\nChange-Id: Ief63f88ba323e3f20300b8ff6a5148a78413ac3d\n(cherry picked from commit fd43d18c0f83c3401c7908fa60f160e4ae7ab4a2\nwith modifications to accomodate for password_hash function later\nremoved via https://review.opendev.org/c/openstack/puppet-openstacklib/+/728595)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/95799c3734bf0287ca809af12c221a4807451401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/95799c3734bf0287ca809af12c221a4807451401"}]},"branch":"refs/heads/stable/ussuri"},"71b25d083e97caa32f0ab6c7adef4fb9344040c1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-30 13:14:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/781756/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/56/781756/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/56/781756/2"}}},"commit":{"parents":[{"commit":"6f32bad5d0876a4661e644198a54a625f39942af","subject":"[Ussuri and before] Add support to specify PostgreSQL database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6f32bad5d0876a4661e644198a54a625f39942af"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-17 10:34:27.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-30 13:14:06.000000000","tz":0},"subject":"Add designate::db::postgresql class.","message":"Add designate::db::postgresql class.\n\nThis allows Designate to be used with PostgreSQL.\n\nBackport note:\nThis backport includes modifications to accomodate for password_hash\nfunction removed during Victoria cycle[1].\n\n[1] https://review.opendev.org/728595\n\nChange-Id: Ief63f88ba323e3f20300b8ff6a5148a78413ac3d\n(cherry picked from commit fd43d18c0f83c3401c7908fa60f160e4ae7ab4a2)\n(cherry picked from commit 47e3e4132fbdf046c0d1f8dcfe1dd2b08b73778f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/71b25d083e97caa32f0ab6c7adef4fb9344040c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/71b25d083e97caa32f0ab6c7adef4fb9344040c1"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
