)]}'
{"id":"openstack%2Fpuppet-designate~800740","triplet_id":"openstack%2Fpuppet-designate~master~I6b57a4ea107a59896e39d07146e90bdef65b4d08","project":"openstack/puppet-designate","branch":"master","topic":"designate-proxy-headers","hashtags":[],"change_id":"I6b57a4ea107a59896e39d07146e90bdef65b4d08","subject":"Add support for oslo_middleware/enable_proxy_headers_parsing","status":"MERGED","created":"2021-07-14 02:15:22.000000000","updated":"2021-07-15 18:58:22.000000000","submitted":"2021-07-15 18:56:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800740-designate-proxy-headers","meta_rev_id":"ad708d5c88e55f6142766e56aeb6d5ecc6ff3685","_number":800740,"virtual_id_number":800740,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 18:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-07-14 13:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-07-15 17:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-07-15 17:07:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 02:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-14 13:59:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-15 17:07:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d14b6b459ecaf6b16139b65ffb870cda4fbd7d14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-14 02:15:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18b7adc904762883d0f4f6ab30b740e29aa89078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-14 02:22:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e38b5c93261543f0be0027d93db7d7e5346bc00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 02:51:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f3fdaabad5b4eacb264351646d6f280 : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfe01c276cc442cbb196b62c331cad61 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddf7f0b0bd0241c59387868ed99ee6c4 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/61e2dbcabb0b494f80c46d5c2f28db87 : FAILURE in 10m 44s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/99d0402b4f6f448f8b097a5554c3d535 : FAILURE in 10m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec9e80ee1c344325b265d1ebb1793835 : FAILURE in 7m 06s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c23ac236bcd4f25a8ec52377ff515f9 : FAILURE in 6m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e01249dcf85c420d8170b81617196e8c : FAILURE in 9m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9b91ed088584c4cad174991a7341321 : FAILURE in 27m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f53583200f194c57b81aa424847ff7fd : FAILURE in 14m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/57f0c28b08d74381bd4df7d18cdb1674 : FAILURE in 13m 36s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f86661e268c84354a550fcaa5a0bc13c : FAILURE in 10m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbabea5abc014e97b3396bc9c8d83ff5 : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bfc04a2d7b54bcb93e66733901483e7 : FAILURE in 10m 08s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/414c35e9bc8f4168b85798f7731c4a72 : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/391ec262b6c34e7f99e9ad407ec6659c : FAILURE in 7m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ab0635436fe4b4cb38d7bf7da7fb963 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"240fd722edd85f9ae5b21fd3a93f7eb8ecb2b7cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-14 12:00:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"85befdd7648d75f5e296e33884d71d71a25b5c79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 13:23:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c56fc2a849c748c0a003db48a6c84e9c : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c59bedece2f14b7ab2234f395d3aab56 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c54a093fcd164462b9e5dd8a865b14d8 : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0561234b6bbf49359fc0b90b26e0000e : SUCCESS in 10m 13s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/55404b321423400e9b036bc84f2fe828 : SUCCESS in 11m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee7fd662f470480c9ed8a6fafcabda59 : SUCCESS in 9m 20s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eca36cf8b2e74574bb3df655122a0e82 : SUCCESS in 7m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ca31bc5860549b98849fb188a4f44d2 : SUCCESS in 9m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5300c5eab15149a79f664265d1bd25a2 : FAILURE in 24m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f98afad20a145568dd5ad0bf1e6f4c9 : SUCCESS in 1h 21m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/669f2fb162e34bf7a636c9af7d1d6648 : SUCCESS in 1h 06m 35s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/35b481ca7e0a4dc8a840d58607ae8a5e : SUCCESS in 22m 20s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17f26aa6dd89465db6dcaa20f6ba106c : FAILURE in 13m 43s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a25ca5add69242c09300d9a01ccd1bf0 : SUCCESS in 24m 39s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c87d5fe677ac4246b112170e4fa4d5e9 : FAILURE in 14m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/748cf0f866d2435abd76f3e5186a43fe : FAILURE in 13m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35f559a3bd0b41d7acbb6c65996c4bf6 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b725d4a253df777ed11de1e0c2d5adefaad55f2e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-14 13:59:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e1c953a32d1d10d2d5de0bf4c1ba216a05cc804f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-07-15 17:07:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"18398f0b146030cb31e4458820d630c438057b83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:08:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01340b3fe2d13c379a3ddca44166e786aeab8efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 18:56:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5bc5835fc824d65891f17e028049ed3 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b22ef13404f4f2a9c2622ada0b730f0 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e46dc462b3864d12b6b20977d1050a92 : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8529ef1fec134609b7802029ad041f66 : SUCCESS in 11m 40s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/0dd424c1691c4626a84bc1ecc014ef40 : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ecab0a0a12047269a68a0912aa0efcc : SUCCESS in 10m 39s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3cc7e0cad254bffbcdca9940b30cf54 : SUCCESS in 8m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91a801d9b626423494d4054c88d397a6 : SUCCESS in 1h 42m 14s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/940671745c2d4d87bd10bd483d56c0bb : SUCCESS in 24m 22s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b59fab515f8401982bf38d4ee999bc2 : FAILURE in 18m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0427e5aa810484eb08c4575f7eb13f7 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"e23b7ae76110e5a802aea4be80e1c5780c16ebb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 18:56:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ad708d5c88e55f6142766e56aeb6d5ecc6ff3685","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 18:58:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65251cf37c89421095a62ab82a9cb09d : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5c7d7e6d58fcfb240673420bc196a75ab888d72c","revisions":{"cfc802c1b4d8a894d2be4248c86603622976f8d1":{"kind":"REWORK","_number":1,"created":"2021-07-14 02:15:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/800740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/800740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/1"}}},"commit":{"parents":[{"commit":"a1f2605c4edf8cc0cc6ed97e2ad3842e6a576210","subject":"Remove leftover of v1 api support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a1f2605c4edf8cc0cc6ed97e2ad3842e6a576210"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-14 02:15:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-14 02:15:02.000000000","tz":540},"subject":"Add support for oslo_middleware/enable_proxy_headers_parsing","message":"Add support for oslo_middleware/enable_proxy_headers_parsing\n\nChange-Id: I6b57a4ea107a59896e39d07146e90bdef65b4d08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/cfc802c1b4d8a894d2be4248c86603622976f8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/cfc802c1b4d8a894d2be4248c86603622976f8d1"}]},"branch":"refs/heads/master"},"d6596e24ee5a06f3464ff04a6da7c87ab293b5e8":{"kind":"REWORK","_number":2,"created":"2021-07-14 02:22:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/800740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/800740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/2"}}},"commit":{"parents":[{"commit":"a1f2605c4edf8cc0cc6ed97e2ad3842e6a576210","subject":"Remove leftover of v1 api support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a1f2605c4edf8cc0cc6ed97e2ad3842e6a576210"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-14 02:15:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-14 02:22:24.000000000","tz":540},"subject":"Add support for oslo_middleware/enable_proxy_headers_parsing","message":"Add support for oslo_middleware/enable_proxy_headers_parsing\n\nChange-Id: I6b57a4ea107a59896e39d07146e90bdef65b4d08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d6596e24ee5a06f3464ff04a6da7c87ab293b5e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d6596e24ee5a06f3464ff04a6da7c87ab293b5e8"}]},"branch":"refs/heads/master"},"5c7d7e6d58fcfb240673420bc196a75ab888d72c":{"kind":"REWORK","_number":3,"created":"2021-07-14 12:00:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/800740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/800740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/800740/3"}}},"commit":{"parents":[{"commit":"a1f2605c4edf8cc0cc6ed97e2ad3842e6a576210","subject":"Remove leftover of v1 api support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a1f2605c4edf8cc0cc6ed97e2ad3842e6a576210"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-14 02:15:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-14 12:00:23.000000000","tz":540},"subject":"Add support for oslo_middleware/enable_proxy_headers_parsing","message":"Add support for oslo_middleware/enable_proxy_headers_parsing\n\nChange-Id: I6b57a4ea107a59896e39d07146e90bdef65b4d08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5c7d7e6d58fcfb240673420bc196a75ab888d72c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5c7d7e6d58fcfb240673420bc196a75ab888d72c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
