)]}'
{"id":"openstack%2Fpuppet-designate~811040","triplet_id":"openstack%2Fpuppet-designate~master~I13f47eb9068a023384817b36e94fc687f1b378cc","project":"openstack/puppet-designate","branch":"master","topic":"auth-ut","hashtags":[],"change_id":"I13f47eb9068a023384817b36e94fc687f1b378cc","subject":"Do not test detail of keystone::resource::service_identity","status":"MERGED","created":"2021-09-26 23:46:58.000000000","updated":"2021-09-27 11:24:07.000000000","submitted":"2021-09-27 11:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811040-auth-ut","meta_rev_id":"21299c0638c0d0dc3baf9e07e27dfc7eb79b0419","_number":811040,"virtual_id_number":811040,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 11:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-27 11:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-09-27 07:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-09-27 11:06:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-27 00:01:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 07:23:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-09-27 11:06:54.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"ed35911429f655be5f3608a0827d804b327b25b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-26 23:46:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51828793ca0c85d58fcb7312ee203567d7644dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 00:01:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0254242e7ddc46279b1bb4de581b82c5 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bae7f2c076c420f81b4785227d6c298 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/242a47a3588645d8a6cd4dbae42783b7 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/038545aebb5e49b0906b7045f193eef1 : FAILURE in 12m 05s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/73f80d9cf8654b67ab8f1a088855f311 : FAILURE in 12m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83442c93962140f8935aa9080ca58cdf : FAILURE in 9m 52s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/025920032f6249a78359e9e0566f6e0d : FAILURE in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33b53c81dbb34f24a38e8a3e44c73bfa : FAILURE in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c844402165ddcd53f79e4a850be0ff92f76cc74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-27 00:15:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e054a793fd1566adfdec22d7e939571a6490725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-27 00:21:42.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7301b1fd23caf6839154685b2a8066129ef7a40b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 00:35:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bf884288d364b958bdb63c147da339e : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9652e07d199e4559842a07d159fe4ca8 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7a55c29e37e43d8904a986098b7aa7f : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/47d12d6c5783481dbfb52263e9fd00ff : SUCCESS in 11m 36s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdb36d761a1e42d8a6b40b072489bb6c : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e649f0e6773945adbb0953a847db34ef : SUCCESS in 8m 22s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91e53d7213974c72b57100984538ff2a : SUCCESS in 8m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5aceaefb142463dbfe5a93fb01ac85e : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b31e4237b1e91f07295ea3e6fadf0d721aa1479c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-27 00:47:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"518ae9838c2d96c3e6995eb43ddb208ce5b05239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-27 00:55:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ff5a1b91c69df8aee7d5f6688b30cf77f50be961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 01:10:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfe56744ac204151b96132c224ba8113 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fb052a58f874d319350389a1d762394 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/790be090318f41569955e43562d07c53 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d079343c48bb421fba5fe51294c11a35 : SUCCESS in 10m 23s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc356de675df4ef581bdc2aa81745d00 : SUCCESS in 10m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/981cc8e474d74ab19c1b45ef8d836348 : SUCCESS in 7m 26s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3fe68f5eff44fc091502c9cfc5c0833 : SUCCESS in 9m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5559f7f8308a45b282206bac8198db09 : SUCCESS in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff319034e9f5c861aa961572495c3412d2ccf855","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-27 07:23:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e2d3a1521fbecba2df1767f1c668e7636e616f68","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-27 11:06:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"75b7bde4b1c404e6361013971e3822b05556520d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:07:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dd9fbb1501d61a5e6090d9349be5dce191e08751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:24:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/964af7087c274dd8b20971093dd94d5a : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2989ccc06604e86bf91c7fa97b0081c : SUCCESS in 4m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b55834a903c5473d86806eead5fe3e07 : SUCCESS in 4m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/99d0d085ca0844ba9e6f33127d3acc06 : SUCCESS in 12m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/58f3478ffd1f43dc8ba4eacd8184a9a7 : SUCCESS in 13m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c84d4e1a974e4ab4815e63cbaf5d9833 : SUCCESS in 10m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1601566a44ff42bfa011e70879d458af : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":5},{"id":"21299c0638c0d0dc3baf9e07e27dfc7eb79b0419","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:24:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3e48d3f6dbe1898ab8f78f1e0b79c8556308ad25","revisions":{"1cdcf5db05c73d0ef5d3bd4f77c91882bb7fc71b":{"kind":"REWORK","_number":1,"created":"2021-09-26 23:46:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/811040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/811040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/1"}}},"commit":{"parents":[{"commit":"9d16312c96295cdbc8f339c6429135d6e317e8d9","subject":"Merge \"Allow purging policy files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/9d16312c96295cdbc8f339c6429135d6e317e8d9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-26 23:46:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-26 23:46:39.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: I13f47eb9068a023384817b36e94fc687f1b378cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1cdcf5db05c73d0ef5d3bd4f77c91882bb7fc71b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1cdcf5db05c73d0ef5d3bd4f77c91882bb7fc71b"}]},"branch":"refs/heads/master"},"a05162f9fcf7654d339bf7598fae7917b9c0ee1f":{"kind":"REWORK","_number":2,"created":"2021-09-27 00:15:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/811040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/811040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/2"}}},"commit":{"parents":[{"commit":"9d16312c96295cdbc8f339c6429135d6e317e8d9","subject":"Merge \"Allow purging policy files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/9d16312c96295cdbc8f339c6429135d6e317e8d9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-26 23:46:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 00:15:00.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: I13f47eb9068a023384817b36e94fc687f1b378cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a05162f9fcf7654d339bf7598fae7917b9c0ee1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a05162f9fcf7654d339bf7598fae7917b9c0ee1f"}]},"branch":"refs/heads/master"},"1a53f5f0fe6998c852779c466ab7c81129fed65a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-27 00:21:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/811040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/811040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/3"}}},"commit":{"parents":[{"commit":"9d16312c96295cdbc8f339c6429135d6e317e8d9","subject":"Merge \"Allow purging policy files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/9d16312c96295cdbc8f339c6429135d6e317e8d9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-26 23:46:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 00:21:26.000000000","tz":540},"subject":"Do not test detail of keystone::resource::service_identity","message":"Do not test detail of keystone::resource::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: I13f47eb9068a023384817b36e94fc687f1b378cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1a53f5f0fe6998c852779c466ab7c81129fed65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1a53f5f0fe6998c852779c466ab7c81129fed65a"}]},"branch":"refs/heads/master"},"c55947153acc67d122cc1ccdc6e58fb47438cdcb":{"kind":"REWORK","_number":4,"created":"2021-09-27 00:47:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/811040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/811040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/4"}}},"commit":{"parents":[{"commit":"f39dfcc437ac0095e61edcdde16df258ac8144fd","subject":"Use consistent spelling, OpenStack, instead of Openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f39dfcc437ac0095e61edcdde16df258ac8144fd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 00:47:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 00:47:04.000000000","tz":540},"subject":"Do not test detail of keystone::resource::service_identity","message":"Do not test detail of keystone::resource::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: I13f47eb9068a023384817b36e94fc687f1b378cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c55947153acc67d122cc1ccdc6e58fb47438cdcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c55947153acc67d122cc1ccdc6e58fb47438cdcb"}]},"branch":"refs/heads/master"},"3e48d3f6dbe1898ab8f78f1e0b79c8556308ad25":{"kind":"REWORK","_number":5,"created":"2021-09-27 00:55:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/811040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/40/811040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/40/811040/5"}}},"commit":{"parents":[{"commit":"df86f122754b084c007575ab9361be4465697eb2","subject":"Fix stdlib version in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/df86f122754b084c007575ab9361be4465697eb2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-26 23:46:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 00:55:02.000000000","tz":540},"subject":"Do not test detail of keystone::resource::service_identity","message":"Do not test detail of keystone::resource::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: I13f47eb9068a023384817b36e94fc687f1b378cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3e48d3f6dbe1898ab8f78f1e0b79c8556308ad25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3e48d3f6dbe1898ab8f78f1e0b79c8556308ad25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
