)]}'
{"id":"openstack%2Fpuppet-designate~816200","triplet_id":"openstack%2Fpuppet-designate~master~I2fc87d6a4a166779e34210a377c99b313a0f1b97","project":"openstack/puppet-designate","branch":"master","topic":"designate-wsgi","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-03 09:50:29.000000000","reason":"Takashi Kajinami replied on the change"},"6681":{"account":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"last_update":"2021-11-03 12:56:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2fc87d6a4a166779e34210a377c99b313a0f1b97","subject":"Add Designate API WSGI support","status":"MERGED","created":"2021-11-01 17:53:10.000000000","updated":"2021-11-03 12:58:26.000000000","submitted":"2021-11-03 12:56:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"816200-designate-wsgi","meta_rev_id":"5e6eef9a17becf335311338b81318f876bb8c1a5","_number":816200,"virtual_id_number":816200,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-03 12:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-11-02 18:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-11-02 19:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-11-02 19:19:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-01 18:43:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 05:16:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-11-02 18:16:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-02 19:19:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a343b15cedfd1e8d79d97483969111cfa8bf6dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-11-01 17:53:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"417fee8493928c7c66d34ca5e239a4114695b8f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 18:43:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31584d560c6e4cda90b00b9f494db493 : FAILURE in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5238806696a14264b06257ac376321a2 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0d58a583f7949c7a5cc6b8e3b38f362 : SUCCESS in 4m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/588c2840e335449e8c76687d1df3d6fc : FAILURE in 12m 22s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0145cb5f80d4a69a387709059b04865 : FAILURE in 11m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c4fff91fed49b385e74b2814c0e388 : FAILURE in 9m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b5ea9c0fc6848568071a3e072d12338 : FAILURE in 9m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d583501b536f4012b5dec9390b5f725c : FAILURE in 9m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8ea17eda24d4feeb5324aeeb0ebf7f6 : FAILURE in 31m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f97e9369b454518a42dd464061fe818 : FAILURE in 46m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/820af0484c01422b949a1977e9624f1b : FAILURE in 15m 38s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b28f5f46bf2c4de29516ad5c3b3cb6be : FAILURE in 21m 59s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abd1d28dd882463daea6c618441b3b00 : FAILURE in 15m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cb15cef580b4f57adee981750dfc702 : FAILURE in 15m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b8629deec6145919725d00fb3fff27d : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7754ce20b8a80f54861bcd89adb44379062df4d1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 05:16:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5b5d3d68bd30b17ab7ffb0d7e6adc5e1a28ed4a5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 05:21:34.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"87591df8a548eb357a680b2f4e29485cffe1bf2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 05:30:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"839afe56b7a0749a680c57fb1f482527dde7baea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 05:52:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d4b5b22f4c52e5bc0798c8a954dcceab981d8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 06:55:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb0fb43374aa46569aa8591ce246737e : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/712aba3a6e4a4a44a06d43aee9e4d8a9 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10f462ebef4241739a78f6e1ac22b4a1 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/81aa34903411456e9166653c045c6ac9 : FAILURE in 9m 43s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3a3f7a033f74cf6bf9769477853e3ee : FAILURE in 9m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2ab04e7eaa543b79dbd252439759112 : FAILURE in 12m 40s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d947b7eb2f24e208376df3d4ebed378 : FAILURE in 8m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f59b14876d234ceebea9d23cd6555b85 : FAILURE in 7m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cc4400aa65a4774971240de58a58ef0 : FAILURE in 29m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c56cbe92a70d472b89a44b4ca833db85 : SUCCESS in 1h 01m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87254699218b4d67ab5c8e47a2f43608 : SUCCESS in 14m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bb51dc18f1d45d88fecf555292d43f6 : SUCCESS in 17m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddbba973887a4c0ab21ea297504fa9e3 : SUCCESS in 14m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d42033c0290248fe80a602b63143ea8d : SUCCESS in 13m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2394ce5e697d4705b7102dc6191278e6 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f7d4776a8b0dbbfe1a6065a7b89b9d5e3980caab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 07:12:35.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"f0fe7ede13291b81d5b73be516bc805ef704c390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 07:31:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1fe4d729647ffa8051d9c932902d78366b3f2ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 07:59:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"425093f7f0754a7e267e73385397ee743e71da83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 08:14:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1f80351fb63f3efe34cf794cadcce401a3c75c7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 08:38:17.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"30c2e84ebe2aa00872bc799129d3e83b6e8e8a10","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 09:10:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e8d55587dd3332021f64fd72d1a76cc5b25073c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 09:11:04.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2d529fbce132c06fb6cd8b01b474ab37fac188d9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 09:17:07.000000000","message":"Topic set to designate-wsgi","accounts_in_message":[],"_revision_number":8},{"id":"5cc58359a3de25cc2229910f58f016a26dc9835e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 10:29:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e419be8480b44e93b147d0efd42a3158 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2280aa27374e4d31b16c5836341995a6 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78796a0c3ceb4374b123a32d23e5ea28 : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9706dd1fa84407586e0ddf38626d879 : SUCCESS in 11m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7336ccdd02d14c1bab1aa536f2f04f3e : SUCCESS in 10m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/273fd399504a450eaa99a6bcc20cd9a7 : SUCCESS in 8m 16s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb4fce69ad3e480fa1f1296d43c823ac : SUCCESS in 8m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0129a61bf5274ff2bff269122c11a215 : SUCCESS in 8m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a54d27ce5454774afbf871ec3a39bc4 : FAILURE in 1h 05m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fc5a5cac36e42d69a8c4966c25f8326 : SUCCESS in 1h 45m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ca882ef3b9c44fda5825d1c4332fe89 : SUCCESS in 15m 29s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4deb75f0aff4abd848b45319767fa20 : SUCCESS in 22m 37s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c92b1bd0bb214725adce5d1237c87df7 : SUCCESS in 14m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/439ab1f329e240a7938dfa714b5780f9 : SUCCESS in 15m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9198400af3124a2ea6399222be2a9eb6 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":8},{"id":"ec540826b22e90f45047329b44261c915473f25d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 13:31:36.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"aba0504192c04b182afd15989b2b7d1b54d34135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 13:35:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"acd236af7871565241a6c74511ade97f4fd28154","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 13:52:24.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a1d2dc0754e76f90766a3994d6936704d475d3bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 13:57:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"68edfe3c809d1083c503a17426fd85238f1839eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 16:39:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ffcfd95354a45e2b3894d45979c9c04 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cee6c0ff120c4896bba7ad942b896923 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/647a0c03afb4487c87a2592366fc0b2c : SUCCESS in 4m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebde3edbaddc4bd397808773d62f44af : SUCCESS in 12m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7c184ee02ff448c82ebce9fa52cf621 : SUCCESS in 14m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3860372a064c4aae99b4367f9a722a63 : SUCCESS in 9m 27s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfdce794b7b249c3b3d1146979b23ac1 : SUCCESS in 9m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a1271565bf54ddbbc950575383f0c73 : SUCCESS in 9m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8231d3a2e7904519af877e99906a28e8 : FAILURE in 34m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea87e7218a44476ba892da96463fd9b4 : SUCCESS in 1h 19m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e07ff15e971c41e7989f85b665fed805 : SUCCESS in 17m 26s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/450507c3504f45b19be9c15573a2395f : SUCCESS in 24m 32s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8714348392f2477289553d1f28d2a954 : SUCCESS in 18m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bcdec2b90044575817a16bb3404c793 : SUCCESS in 19m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b6d058f205a4ecfba0b76fe4a2c20e2 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":10},{"id":"54b3b04824c977143fddb9f1ea87af26e79974c4","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-11-02 17:13:02.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"80a53f82b2bfa0eedbba6c0b1cd07364eebcfa40","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 18:16:23.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7cf0492b8a6b02e665906ded0d5a7a70a1462703","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-02 19:19:16.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"59cbeb308733b3310e82dc75976fde54d2823c96","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 09:50:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d23e12fbc007e56a583c5afda3425fa848fa2220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:27:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2564953725d74b5fb9a511362c5ec315 : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8b0ff9ecf4048ff89c37c6d6452465a : SUCCESS in 4m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f195c6f65204d048f75049fee5a8ca4 : SUCCESS in 4m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4e8ac0aa75a41c3a353ead60f868827 : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f928d45b819b486aab59272cf4764ae6 : SUCCESS in 12m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/110f4092f8204543a0507cdf9b8646e7 : SUCCESS in 11m 46s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d178432b1d784439b7e425ae567bd9e2 : SUCCESS in 9m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f9ccc7aaf184eacbf8c965052b392d5 : SUCCESS in 9m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b51b3da9dae84ccaab2f2a2bbea3d110 : FAILURE in 58m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f35b7127fc74b4ab79eb366887b89c2 : SUCCESS in 1h 28m 30s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/358b4f78915b435a9bf7216a0d8bef66 : SUCCESS in 15m 32s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/422069068fab4f34bbb1d27d33d56dfc : SUCCESS in 23m 07s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/564e539dea8d4dbc88c667dfd7d83eaa : SUCCESS in 16m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0519b1367b0d42188d5a18d18d6a2a29 : SUCCESS in 15m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e2c4192a5ca43d2870976d462511f09 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":10},{"id":"ddc0644e65840b3ac615593efec8fc8f702e648e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:28:59.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4362b7ea8c7ae0a902eeb187af03232b3d7092cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:56:27.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/716f3a19944a43dfb701b6f2fda53bd7 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/860d1f2af30740d8be798dd02dc8f287 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c104890df5e44acbf241c57feebc03b : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b4c846a9f7d4b8cb469f39c658cfc1a : SUCCESS in 11m 34s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/feee2dd1d9c947ecad9ece164aa878ba : SUCCESS in 11m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bdc475baf484e328e6dd804ab07d424 : SUCCESS in 8m 45s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6be118ab46a4009b0239eb40565d085 : SUCCESS in 8m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/907c7360c23343f2be6693aead3b3aef : SUCCESS in 1h 22m 02s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cebc600b3947402d9083bfcec48a327d : SUCCESS in 22m 08s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4071195fda647c6b087d02f518d0c7e : SUCCESS in 14m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e62c4dc583764f45aff89e539a1cd0c0 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":10},{"id":"bd1562092b77f0924c21c322f930fd1bd7b9422c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:56:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"5e6eef9a17becf335311338b81318f876bb8c1a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:58:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9de83235a8ad495c8e3cf34333aa7fb3 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"1288b062e8f0ac88eea9eb6e83ba6cffd119c982","revisions":{"9dce4746716f5166806573fb1680aeb33a556738":{"kind":"REWORK","_number":1,"created":"2021-11-01 17:53:10.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/00/816200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/1"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:53:04.000000000","tz":-150},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/9dce4746716f5166806573fb1680aeb33a556738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/9dce4746716f5166806573fb1680aeb33a556738"}]},"branch":"refs/heads/master"},"71d3129613230cd1c3f4ab5b2d8e639bb3eeb2f1":{"kind":"REWORK","_number":2,"created":"2021-11-02 05:30:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/2"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 05:30:17.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/71d3129613230cd1c3f4ab5b2d8e639bb3eeb2f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/71d3129613230cd1c3f4ab5b2d8e639bb3eeb2f1"}]},"branch":"refs/heads/master"},"c35c2f3186c102b7fe3204f355f5a2fcd5a5950d":{"kind":"REWORK","_number":3,"created":"2021-11-02 05:52:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/3"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 05:51:53.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c35c2f3186c102b7fe3204f355f5a2fcd5a5950d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c35c2f3186c102b7fe3204f355f5a2fcd5a5950d"}]},"branch":"refs/heads/master"},"da8ebf26d2a5bdfc770bebacfd82537b419b996b":{"kind":"REWORK","_number":4,"created":"2021-11-02 07:12:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/4"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 07:12:35.000000000","tz":0},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da8ebf26d2a5bdfc770bebacfd82537b419b996b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da8ebf26d2a5bdfc770bebacfd82537b419b996b"}]},"branch":"refs/heads/master"},"f7e556d3ac80b2ca13e5c7f8e20691b4b460cda6":{"kind":"REWORK","_number":5,"created":"2021-11-02 07:31:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/5"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 07:31:06.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f7e556d3ac80b2ca13e5c7f8e20691b4b460cda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f7e556d3ac80b2ca13e5c7f8e20691b4b460cda6"}]},"branch":"refs/heads/master"},"c90bce12dab7df3b4170331932ee64c316ff6f6d":{"kind":"REWORK","_number":6,"created":"2021-11-02 07:59:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/6"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 07:58:52.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c90bce12dab7df3b4170331932ee64c316ff6f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c90bce12dab7df3b4170331932ee64c316ff6f6d"}]},"branch":"refs/heads/master"},"9b807da7d2b5648756ec61dcd982ec6222b5a7d6":{"kind":"REWORK","_number":7,"created":"2021-11-02 08:14:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/7"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 08:14:50.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/9b807da7d2b5648756ec61dcd982ec6222b5a7d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/9b807da7d2b5648756ec61dcd982ec6222b5a7d6"}]},"branch":"refs/heads/master"},"845e032f7f7d9fd822ef5397acfb5ac2ab9738b5":{"kind":"REWORK","_number":8,"created":"2021-11-02 08:38:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/8"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 08:38:09.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/845e032f7f7d9fd822ef5397acfb5ac2ab9738b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/845e032f7f7d9fd822ef5397acfb5ac2ab9738b5"}]},"branch":"refs/heads/master"},"0582931e0f8f240e42a9ad63703161fbedd1b3b1":{"kind":"REWORK","_number":9,"created":"2021-11-02 13:35:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/9"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 13:34:54.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0582931e0f8f240e42a9ad63703161fbedd1b3b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0582931e0f8f240e42a9ad63703161fbedd1b3b1"}]},"branch":"refs/heads/master"},"1288b062e8f0ac88eea9eb6e83ba6cffd119c982":{"kind":"REWORK","_number":10,"created":"2021-11-02 13:57:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/816200/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/00/816200/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/00/816200/10"}}},"commit":{"parents":[{"commit":"2e4b2c046bf3b9c5a4306e03f547ec14d025c579","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2e4b2c046bf3b9c5a4306e03f547ec14d025c579"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 13:57:36.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1288b062e8f0ac88eea9eb6e83ba6cffd119c982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1288b062e8f0ac88eea9eb6e83ba6cffd119c982"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
