)]}'
{"id":"openstack%2Fpuppet-designate~819274","triplet_id":"openstack%2Fpuppet-designate~master~Id0ba4c95005d148477a313f0aa5edddc3c681e15","project":"openstack/puppet-designate","branch":"master","topic":"system-scope-params","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-07 23:49:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0ba4c95005d148477a313f0aa5edddc3c681e15","subject":"Accept system scope credentials for Keystone API request","status":"MERGED","created":"2021-11-25 09:14:00.000000000","updated":"2022-01-07 23:51:39.000000000","submitted":"2022-01-07 23:49:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819274-system-scope-params","meta_rev_id":"76f1206a339b06c61efed911ca3a19faa0d810aa","_number":819274,"virtual_id_number":819274,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 23:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-12-13 15:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-07 22:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-07 22:16:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 10:22:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 15:34:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 22:16:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a8b08e192734e508e7b576c0f12ef4c713a576b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 09:14:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"866f42f67a9a9cf205e2c600f6153dc2bcf5266d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 10:22:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf979c27ad34e48902b8f8db81c05b8 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4305b6b682054610a1704830f1f859fe : SUCCESS in 4m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4818b09ec899464d807244dc881055b7 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cccd4b20f6364fa886d79030f5a3e370 : SUCCESS in 11m 25s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/105627d5800b4e49bf81bee17294f035 : SUCCESS in 13m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af01004b022842f389d83e73421ddd99 : SUCCESS in 10m 01s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3390a3fb8ba9413089b4bee4432126fa : SUCCESS in 8m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e7510af002c4c5397c26e85159e2d82 : SUCCESS in 10m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df79440ffaa04a1183c058bfc9c052d0 : FAILURE in 1h 02m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/614e06cc6b8c4d089d87c688d96dbd5f : SUCCESS in 1h 06m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3b72fb83c5e4a2da116b05313fb1922 : SUCCESS in 15m 58s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2c86c18897d4f3aadc05e873b6e4cb8 : SUCCESS in 19m 16s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64415e46f82f40c9a53745f7d7bed749 : SUCCESS in 17m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ff9496339f94377b078638b1c658fa2 : SUCCESS in 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d46b2d5084517b2f823f89595ddeac1c8629be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 11:31:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f20aab4887755b196c3679e319988bea080aee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 11:41:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c0941acaaa425c8d82d9ba167e2a9a9ea99eb48b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:17:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e3cd95cc21c4a369065776c34caa035 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4ca6568744b41dc94d27b02f3ffef29 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/948b356a289b462a91c363f29d06d00f : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/46f8c254af7549ddbc464b022a59ced1 : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/684040192cef44589b5b8d1459d7deb8 : SUCCESS in 12m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557875d16a3547bc8a4ab01adb1254eb : SUCCESS in 9m 01s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f2d741fb1064297be1b06bdd95fcbad : SUCCESS in 8m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02d49204997d4dffab19db8ffb72e37d : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cc909346cd54bf195cd17efdf291d00 : FAILURE in 28m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4bc5a1c14934af092fa7c1def65ddd4 : SUCCESS in 1h 28m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c148788b47c74faab302d661590388e9 : SUCCESS in 15m 55s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9b845f4593d453483c46638177ee5da : SUCCESS in 23m 51s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff3fa21b66724eabbd7189fd0ded1dad : SUCCESS in 15m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d3323abd1cb4596a9976e23a427964b : SUCCESS in 16m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d7055061d3a4e11bfee4c7302f889aa : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"1a3755fe3a8f8f0894623fec41d8e23ee335fa1d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-13 15:34:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d6e19e5cdb35cb15dddbe83eca74a2570803d1d4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-07 22:16:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"468670a73efbadccd269e91e43cadf14d9dd1014","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:18:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e019894254fa5a69472391b4db03da1f60b1bae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:49:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92b7dada2ec9474d9d1e09ff94ad84e0 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/987cfe5c1ca24855b361013cdc8102e9 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61f0678e6e514e819b98a324a8206bd4 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6b3bf6cad124705bc0c8a62e01625f4 : SUCCESS in 13m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6403e6ff6614dc0b15a5d4dbfc87e4d : SUCCESS in 13m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36ecac4659474042a494a2cf9787ce74 : SUCCESS in 11m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6661f3d85c44d4c8e50072c6d3a26a9 : SUCCESS in 9m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/501d741a5ce7499eae0444a3b853e911 : SUCCESS in 1h 26m 48s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/da31036bc6274366b0cbc19a6b5f5fc9 : SUCCESS in 24m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9ed7aa597a54a6f9e18718c13d3b06b : SUCCESS in 19m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae0ddbe6a70a48f8bde14e0ba46671f2 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e6c0c5e91acafaf8ff95b9331346ec287231ebb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:49:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"76f1206a339b06c61efed911ca3a19faa0d810aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:51:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1942c242c604c4dbf79d707a2cb97be : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d95528c4bf43a9ed5dc74675af86611f4cceaf41","revisions":{"e61f5c32d4149790cba348f94863641b4238f196":{"kind":"REWORK","_number":1,"created":"2021-11-25 09:14:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/819274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/74/819274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/1"}}},"commit":{"parents":[{"commit":"e1cba4a00288b6a8ad9ffb7d51949010ff5def24","subject":"Use consistent spelling, OpenStack, instead of Openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e1cba4a00288b6a8ad9ffb7d51949010ff5def24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 09:13:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 09:13:28.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following two items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: Id0ba4c95005d148477a313f0aa5edddc3c681e15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e61f5c32d4149790cba348f94863641b4238f196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e61f5c32d4149790cba348f94863641b4238f196"}]},"branch":"refs/heads/master"},"bf33df464cf30dd72c4ea8d9de9b55a80a3678f1":{"kind":"REWORK","_number":2,"created":"2021-11-25 11:31:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/819274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/74/819274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/2"}}},"commit":{"parents":[{"commit":"e1cba4a00288b6a8ad9ffb7d51949010ff5def24","subject":"Use consistent spelling, OpenStack, instead of Openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e1cba4a00288b6a8ad9ffb7d51949010ff5def24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 09:13:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 11:31:20.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following two items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: Id0ba4c95005d148477a313f0aa5edddc3c681e15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/bf33df464cf30dd72c4ea8d9de9b55a80a3678f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/bf33df464cf30dd72c4ea8d9de9b55a80a3678f1"}]},"branch":"refs/heads/master"},"d95528c4bf43a9ed5dc74675af86611f4cceaf41":{"kind":"REWORK","_number":3,"created":"2021-11-25 11:41:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/819274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/74/819274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/74/819274/3"}}},"commit":{"parents":[{"commit":"e1cba4a00288b6a8ad9ffb7d51949010ff5def24","subject":"Use consistent spelling, OpenStack, instead of Openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e1cba4a00288b6a8ad9ffb7d51949010ff5def24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 09:13:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 11:41:09.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following two items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: Id0ba4c95005d148477a313f0aa5edddc3c681e15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d95528c4bf43a9ed5dc74675af86611f4cceaf41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d95528c4bf43a9ed5dc74675af86611f4cceaf41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
