)]}'
{"id":"openstack%2Fpuppet-designate~819972","triplet_id":"openstack%2Fpuppet-designate~master~I80d14d57d75aaa228f009eba1d25d79f24603307","project":"openstack/puppet-designate","branch":"master","topic":"prep-release-yoga-m2","attention_set":{},"removed_from_attention_set":{"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2022-01-07 14:38:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I80d14d57d75aaa228f009eba1d25d79f24603307","subject":"Prepare Yoga M2","status":"MERGED","created":"2021-12-01 06:43:38.000000000","updated":"2022-01-07 14:38:15.000000000","submitted":"2022-01-07 14:38:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819972-prep-release-yoga-m2","meta_rev_id":"72bccc9c3592c9b70b2b7bfcf12c6cf1c27fc8b4","_number":819972,"virtual_id_number":819972,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 14:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-01-07 14:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-01-07 14:38:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-01-07 14:18:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-01 06:52:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 15:34:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 14:18:42.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"fb95c21c9acd890c79d8dfd4ca718ce6078406a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-01 06:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d242dbad0325632bb458558de01ad4bdbadb703b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 06:52:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f7dc911b18be4de0a34f4c161dfce211 : SUCCESS in 2m 22s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d76a7ff2ad924d3490e7feaa26116a15 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fc5bfaaf10f227e6aca35ab9f5600c0dc2b454cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-03 08:56:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0ed52edb1e3811c00341da2f109965533a098b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 09:03:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/985703159baa44179d27ae4a7eca64f1 : SUCCESS in 2m 14s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2677555ef4041a788019bc938986b90 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3b8dd98cfcf5344f76c2a00058bb4ce09c8c607c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-13 15:34:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c600e91bd2af4779a890b9497797ce4ece08253","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:39:57.000000000","message":"Topic prep-release-yoga-m1 removed","accounts_in_message":[],"_revision_number":2},{"id":"7c23b4dec3c891c68a6942dc382537fedec5c00e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:40:00.000000000","message":"Topic set to prep-release-yoga-m2","accounts_in_message":[],"_revision_number":2},{"id":"8649122a3d61b4d93feb4a6dec7fc3f6fe3249fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:18:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"00ec029e16b91ac685f23d65f8b408feefef02a4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:18:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c8d818e4dbda71fffd4588c40101a98b20408034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:29:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/97aeff95bb564426b16d93b5ccf3c5cf : SUCCESS in 2m 40s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d80efcb6c881447cb4b5b0136c354997 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"707568eb4bd7469edf915dd7b8c8f5f0378684bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:30:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"048ba7614a2a957203e8858d520a85bad1b89b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:38:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c266e8d976744b67a5e1d03c3eb1126b : SUCCESS in 2m 40s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2483ad9edfe4d8e93756b7bbefb27e0 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"374d607b96d4268f336360f5a402a1a119a09167","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:38:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"74bae32e894034d0347c11370cb5e3ce26e79e83","revisions":{"8df7453fcff28d94f37a132679cc57987078b3d7":{"kind":"REWORK","_number":1,"created":"2021-12-01 06:43:38.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/72/819972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/72/819972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/1"}}},"commit":{"parents":[{"commit":"a371c5297fa95f6da2c4efdbe50d894672fbf6e8","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a371c5297fa95f6da2c4efdbe50d894672fbf6e8"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"subject":"Prepare Yoga M1","message":"Prepare Yoga M1\n\nUpdate the version metadata for Yoga milestone M1\n\nChange-Id: I80d14d57d75aaa228f009eba1d25d79f24603307\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8df7453fcff28d94f37a132679cc57987078b3d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8df7453fcff28d94f37a132679cc57987078b3d7"}]},"branch":"refs/heads/master"},"18ebd535ace4b6b99a5028556a9fda5ddd6e7f41":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-03 08:56:26.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/72/819972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/72/819972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/2"}}},"commit":{"parents":[{"commit":"a371c5297fa95f6da2c4efdbe50d894672fbf6e8","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/a371c5297fa95f6da2c4efdbe50d894672fbf6e8"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-02 05:59:01.000000000","tz":480},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I80d14d57d75aaa228f009eba1d25d79f24603307\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/18ebd535ace4b6b99a5028556a9fda5ddd6e7f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/18ebd535ace4b6b99a5028556a9fda5ddd6e7f41"}]},"branch":"refs/heads/master"},"74bae32e894034d0347c11370cb5e3ce26e79e83":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-07 14:18:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/819972/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/72/819972/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/72/819972/3"}}},"commit":{"parents":[{"commit":"25d74711b261f3a64e44cf2108faa66528014e85","subject":"Merge \"Accept string for enabled_notification_handlers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/25d74711b261f3a64e44cf2108faa66528014e85"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-07 14:18:37.000000000","tz":0},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I80d14d57d75aaa228f009eba1d25d79f24603307\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/74bae32e894034d0347c11370cb5e3ce26e79e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/74bae32e894034d0347c11370cb5e3ce26e79e83"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
