)]}'
{"id":"openstack%2Fpuppet-designate~821531","triplet_id":"openstack%2Fpuppet-designate~master~Iad8849e77d4139e5473c66f699b5bee640e9dceb","project":"openstack/puppet-designate","branch":"master","topic":"producer_task","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-07 23:49:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad8849e77d4139e5473c66f699b5bee640e9dceb","subject":"Add support for [producer_task:*] options","status":"MERGED","created":"2021-12-13 06:30:23.000000000","updated":"2022-01-07 23:52:43.000000000","submitted":"2022-01-07 23:49:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821531-producer_task","meta_rev_id":"04f2982911ee9fb594a70a25306dcaf081158c5b","_number":821531,"virtual_id_number":821531,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 23:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-12-13 15:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-07 22:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-07 22:20:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 08:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 15:34:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 22:20:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"07d512e001be99b1697d7ba78379e5aafbe4e5eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-13 06:30:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fb7114981d71d3d419f195f80a431e19869f82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-13 07:26:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e4baaf28b79f65277e053afa5d871bbf8827d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 08:42:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43e41ef7f7ad45818ae70503363ab96b : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6012fddfeddc4806af16270c9fba3b58 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d358bc76274d4eb7a7a4bb193993ffd6 : SUCCESS in 2m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/96b5000726d2480caf77c655c98641b3 : FAILURE in 11m 36s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce79bf4e43b84e7cac498d7a85808807 : FAILURE in 12m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37e184665e4f416fb23a975d89c98078 : FAILURE in 8m 03s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0e1bffa68d0429a8d93625fd553085e : FAILURE in 7m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a586f1f44584d45b4deb9daf9dd5ed4 : FAILURE in 7m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ced8533b1b241789e67999443c48a40 : FAILURE in 27m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5da9fded17f84629965f723495638d7f : SUCCESS in 1h 13m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dd9cf54feb541548f1e0c04dad4fd56 : SUCCESS in 48m 11s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29de318940164868973548da513e2558 : SUCCESS in 12m 50s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc79d9bcf69441b7bb0759948b099fc4 : SUCCESS in 20m 45s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8d8679295234b89a76309724b703100 : SUCCESS in 14m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/956ed2c7228b412c971ad43a22f76cdf : SUCCESS in 13m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f512d79676dd48b1bf34c5a056758ebf : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c634fb5420d026659ea2a9a3d4cfa47b739f98a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-13 10:58:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63f1ec5fe859f6d862794a6e890a1a604e2cb5d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 12:18:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfe2548c5e8d45e18001dc2da216ffcf : SUCCESS in 5m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c8f0016dce44b15b83d1636d9455621 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec12bd8b2bac4606b2e09c94366e49be : SUCCESS in 4m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6aeeaa2208684cef82220895fb16a899 : FAILURE in 15m 25s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c68d9b4574704f33bf45b00b92eb9269 : FAILURE in 11m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73473e6c41ab47c5aa9bdf3336a7892d : FAILURE in 11m 14s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8625f992cd954aada1e83aaa7198e15a : FAILURE in 11m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbb15e38721146c79e988d797b2f3d97 : FAILURE in 11m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a969b35b02124186b886999844db1b87 : FAILURE in 1h 03m 18s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f22d1d14d364a688dbd0ee75fd4d82e : SUCCESS in 1h 17m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c60da829c94e4ad28ebe8047b3ae4598 : SUCCESS in 57m 00s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d864a2ebbd74624be524a0efa73cafc : SUCCESS in 17m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3c939bddad64659af2a878892a96f24 : SUCCESS in 21m 49s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2933f271dd0b4b678067578454390b22 : SUCCESS in 13m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66a44f38f9534aada705d839d29406d1 : SUCCESS in 14m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6de3e88a85014f87873f8f617049577b : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"36786987781a5eefa102b6f84be7da0454fbb544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-13 13:09:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7cdadf82be34ccf2675079355ffcf13a4c9c6a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 14:42:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9607eca01711492ca0587c87ffa25c4f : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faa428be563848b0b321032922c3d2ca : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bffaa06d8a4d42efa20de1f1a53410dd : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d38a36d9c644150b1bb87d90dd68dc3 : SUCCESS in 13m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f34c18f401ab476cafe592799494096c : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9397566e7dd2483f8cb4aa6a15742e04 : SUCCESS in 9m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e64d21eef9487a96a921fb26b4e568 : SUCCESS in 11m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a208b0a9f314ad3a28e69b8c33e423f : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d66cc87060a947d29ab2360f9637221b : FAILURE in 59m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce7361571c764fe585bcb1adc95ce4fd : SUCCESS in 1h 12m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4692ccfd0e63462cbdb50884f66f723e : SUCCESS in 1h 17m 31s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c6d63330d4c4546a4dd5763f491cb94 : SUCCESS in 17m 58s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc4aab0290af42199b4e497b2da396cb : SUCCESS in 20m 28s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a518948a26404247b8944aab893fcf4e : SUCCESS in 16m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aef7a7a1b004bda99924eb9d389caee : SUCCESS in 15m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4071be2b9059455f80707fb794417170 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"a341e150c8ba2cd0ab2c18f2c84ca55921ee92aa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-13 15:34:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b82fd999f7c516723e0190e014921947675583e5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-07 22:20:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"796c634ae5674e2f68138df657dda12554667405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:22:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f73b991df42b4a34ee05684fcf48ef64ec7f1758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:49:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cec4326a32e43ca9dbf2be6b113a551 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e16fe3dc17c41caa11dbfdbeb8af996 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2389b025d8a14067b4fc1190d1ec9871 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bf4aa2cd90640a1bb5cf40ce29046c7 : SUCCESS in 12m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/769e3fe9f07c42689fb438941e6ac71e : SUCCESS in 16m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/991a3ee3576f48a1863424cb0cdfdecf : SUCCESS in 10m 49s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a293aa4032e24000844d265cbfb91cf3 : SUCCESS in 10m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/55d9c6c7b1a34d3292f2332f5db7ac9d : SUCCESS in 1h 18m 15s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e0b7c2dee064d7880f02abd15f162e1 : SUCCESS in 22m 02s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b836970af224a108d5d83014be491d1 : SUCCESS in 16m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9463f04b28f54e128d20afa8235715e0 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":4},{"id":"ad9dd33de23a7b613a4b6f72ac060cd256d2a331","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:49:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"04f2982911ee9fb594a70a25306dcaf081158c5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:52:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12bef738057349809b2a65c46c5444fa : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5c53236fc2facb991b3a54a91844a94cc9efa8bc","revisions":{"c159aaa7b81176f53156a4183b1ca8ab124bd335":{"kind":"REWORK","_number":1,"created":"2021-12-13 06:30:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/821531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/31/821531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/1"}}},"commit":{"parents":[{"commit":"8d114592bb302c6e3129386875e74ef039cad079","subject":"Clean up direct dependencies on puppetlabs-apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8d114592bb302c6e3129386875e74ef039cad079"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 06:28:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 06:28:13.000000000","tz":540},"subject":"Add support for [producer_task:*] options","message":"Add support for [producer_task:*] options\n\nThis change introduces support for tunable parameters in the following\nsections.\n\n [producer_task:delayed_notify]\n [producer_task:periodic_exists]\n [producer_task:periodic_secondary_refresh]\n [producer_task:worker_periodic_recovery]\n [producer_task:zone_purge]\n\nChange-Id: Iad8849e77d4139e5473c66f699b5bee640e9dceb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c159aaa7b81176f53156a4183b1ca8ab124bd335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c159aaa7b81176f53156a4183b1ca8ab124bd335"}]},"branch":"refs/heads/master"},"d1fa94e3e87fe3f85719e268a6a55468698861ab":{"kind":"REWORK","_number":2,"created":"2021-12-13 07:26:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/821531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/31/821531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/2"}}},"commit":{"parents":[{"commit":"8d114592bb302c6e3129386875e74ef039cad079","subject":"Clean up direct dependencies on puppetlabs-apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8d114592bb302c6e3129386875e74ef039cad079"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 06:28:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 07:25:52.000000000","tz":540},"subject":"Add support for [producer_task:*] options","message":"Add support for [producer_task:*] options\n\nThis change introduces support for tunable parameters in the following\nsections.\n\n [producer_task:delayed_notify]\n [producer_task:periodic_exists]\n [producer_task:periodic_secondary_refresh]\n [producer_task:worker_periodic_recovery]\n [producer_task:zone_purge]\n\nChange-Id: Iad8849e77d4139e5473c66f699b5bee640e9dceb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d1fa94e3e87fe3f85719e268a6a55468698861ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d1fa94e3e87fe3f85719e268a6a55468698861ab"}]},"branch":"refs/heads/master"},"8afa8dc27d66afa571f42f7f9e50afb9d44fa617":{"kind":"REWORK","_number":3,"created":"2021-12-13 10:58:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/821531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/31/821531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/3"}}},"commit":{"parents":[{"commit":"8d114592bb302c6e3129386875e74ef039cad079","subject":"Clean up direct dependencies on puppetlabs-apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8d114592bb302c6e3129386875e74ef039cad079"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 06:28:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 10:58:01.000000000","tz":540},"subject":"Add support for [producer_task:*] options","message":"Add support for [producer_task:*] options\n\nThis change introduces support for tunable parameters in the following\nsections.\n\n [producer_task:delayed_notify]\n [producer_task:periodic_exists]\n [producer_task:periodic_secondary_refresh]\n [producer_task:worker_periodic_recovery]\n [producer_task:zone_purge]\n\nChange-Id: Iad8849e77d4139e5473c66f699b5bee640e9dceb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8afa8dc27d66afa571f42f7f9e50afb9d44fa617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8afa8dc27d66afa571f42f7f9e50afb9d44fa617"}]},"branch":"refs/heads/master"},"5c53236fc2facb991b3a54a91844a94cc9efa8bc":{"kind":"REWORK","_number":4,"created":"2021-12-13 13:09:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/821531/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/31/821531/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/31/821531/4"}}},"commit":{"parents":[{"commit":"8d114592bb302c6e3129386875e74ef039cad079","subject":"Clean up direct dependencies on puppetlabs-apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8d114592bb302c6e3129386875e74ef039cad079"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 06:28:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 13:09:11.000000000","tz":540},"subject":"Add support for [producer_task:*] options","message":"Add support for [producer_task:*] options\n\nThis change introduces support for tunable parameters in the following\nsections.\n\n [producer_task:delayed_notify]\n [producer_task:periodic_exists]\n [producer_task:periodic_secondary_refresh]\n [producer_task:worker_periodic_recovery]\n [producer_task:zone_purge]\n\nChange-Id: Iad8849e77d4139e5473c66f699b5bee640e9dceb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5c53236fc2facb991b3a54a91844a94cc9efa8bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5c53236fc2facb991b3a54a91844a94cc9efa8bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
