)]}'
{"id":"openstack%2Fpuppet-designate~823704","triplet_id":"openstack%2Fpuppet-designate~stable%2Fwallaby~I2fc87d6a4a166779e34210a377c99b313a0f1b97","project":"openstack/puppet-designate","branch":"stable/wallaby","topic":"designate-apache-wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-07 23:47:02.000000000","reason":"Change was submitted"},"6681":{"account":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"last_update":"2022-01-07 23:47:02.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-01-06 21:50:45.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I2fc87d6a4a166779e34210a377c99b313a0f1b97","subject":"Add Designate API WSGI support","status":"MERGED","created":"2022-01-06 15:08:49.000000000","updated":"2022-01-07 23:48:47.000000000","submitted":"2022-01-07 23:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823704-designate-apache-wallaby","meta_rev_id":"ce7f9511464742590896866be18c2192f67c555a","_number":823704,"virtual_id_number":823704,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 23:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-01-07 06:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-01-07 15:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-01-07 15:05:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 16:38:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 18:50:45.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-01-06 18:50:45.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 22:03:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"}],"messages":[{"id":"38cfd9f0f989e7f8ceb5318d586c39196e464052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2022-01-06 15:08:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa57343413085d1e4b6d0a4eb250cfb262e47fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 16:38:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54ea43d80f2046309beb880563dc2f89 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b408dbbf279b40709876d6802e2c7cdc : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/baceccf0cb71482f80b080e1a5e7f7e4 : SUCCESS in 14m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b031c20d158b4544bf64301d324d481d : SUCCESS in 10m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7612eedcd2894543ba84700f0ac7864c : SUCCESS in 8m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d2e95e202eb4e36bbe51b683b798d75 : FAILURE in 40m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1600b5f3f42240a2bd90c3b039823ba9 : SUCCESS in 1h 24m 04s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/28084cc7cbd349aba3e2ebb0e6170350 : SUCCESS in 21m 50s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fef2cbed43d04197b27e7de1eebca0eb : SUCCESS in 14m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c0a10472e464d09a7ac45187d4143cb : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bc75fcf498d397aab86eb83a6f9b7582008b5201","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-06 21:50:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1379771f2e7c1fefdc903e85d0e0dfbf66ed3346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 06:11:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b0b4026fdc53df0cde3185869a78bb5e4eca28f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 06:12:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69e417bd6dcc74c9fe50d191c3005d1d427d081a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 06:47:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7229da72943b40d9ac20027744ddb4a8 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/098b13fecf9c4683aa6be9d7c6278f15 : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a87b26ed3ec3497dab3c9da9325ea11d : SUCCESS in 12m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0ab22d6f94f448bba13d2f05235fc91 : SUCCESS in 8m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5564b99b95f49d69b05a5d380ed0aca : SUCCESS in 9m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e4fdcdcaff2426ca174a52708381280 : FAILURE in 31m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/397abfe0c80d4c3c947167f52bddd4fe : FAILURE in 14m 23s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/caaccb69d1774830bebd83492b93d22e : SUCCESS in 22m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a98346d5c3a41d5beca79bd3d028fd6 : SUCCESS in 14m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fc6357980b542fcb7755225bed36aa9 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"2fada7a5fb52749ed9b5b8c6f9c413f3d3c2738b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 06:59:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65c3cdc4f4d068b47dc3e70978e717e4f21e785a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 08:55:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe51ffc3c2a64cdc98a221f3e0f55b0d : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44a0524258094f12b923b69224ee92b2 : SUCCESS in 6m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2fedab7577a49a892f18971f3261055 : SUCCESS in 9m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc3c891992a3487bb057a19622c29a98 : SUCCESS in 8m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d86f408dd04cb3a4914d1198268a92 : SUCCESS in 9m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d34a72a8bbe46c0929f3d8631c5215e : FAILURE in 1h 07m 46s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a5baafb6f3c445fa672b572b567c823 : SUCCESS in 1h 49m 04s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cbc215e056841f987f7e7709bc4a46a : SUCCESS in 23m 17s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f07252fc4c42455eb7c7a1c7f6882150 : SUCCESS in 15m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b46c94fa37b4cfb8eccb315d7fefbbe : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"125037c1009d286f9a56dceeb836f03dce7fbcd5","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2022-01-07 13:41:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"369159328d2574dd635c865f7d581ac5a4d08f06","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-07 15:05:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5263f2fb732a39aadb0fd18a34ddffd1f5e70a5d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-07 22:03:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a75a36117f850adfae23e0af6ef03f1c3a97eae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:04:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f66e7724ff89644e82b1308d724ab05f441576cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:35:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/434766e0a8e84f33a9fce1b92b6ffe69 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3835d239f6a34d2d94215878fdbd1393 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d24aacc207e4fafa57832910a6acd39 : SUCCESS in 13m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfcee20854ca4fc7a0504c890510d878 : SUCCESS in 8m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b482c553c5104fdaa659c722e2af6fe4 : SUCCESS in 10m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d227af4410e43e685dfd122012d74b6 : FAILURE in 32m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/68ead74ba8174e15953cccd5a03a18b6 : SUCCESS in 1h 20m 53s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a97585141124602a63016b757e454c9 : SUCCESS in 23m 28s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec7648abe132455084954d55567bdc58 : SUCCESS in 16m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b7e75478f8c48c6ba091bced978aec4 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"7981bbcc0460ce9e17710bca017bfa4565bdc182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:47:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e642a6c635a3424293b3d7f477138331 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdb27dadff974fc092755f843c54ab89 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d680dcddade4efdbcb0fc06174bf1e4 : SUCCESS in 12m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/264cd21675df49e386cc571419ee4573 : SUCCESS in 11m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c89cc388475849139dbbf064d8007a63 : SUCCESS in 1h 37m 14s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f59e7e9858394dfc9f3d262f91fbc700 : SUCCESS in 24m 18s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28513aba4aa64a98961d5a93d54045d1 : SUCCESS in 18m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7fe9cd0b967463287254f355671cb74 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2b88e322d3ceae176104e87e24440949160ce13c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:47:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ce7f9511464742590896866be18c2192f67c555a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:48:47.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e79b5de323642d1a88915b668f2cf75 : FAILURE in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b35ca1f81ab612f42bd3322ccd4a6082f90b17a","revisions":{"34733390cf1d059b59dd5431030a4febfc699d0b":{"kind":"REWORK","_number":1,"created":"2022-01-06 15:08:49.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/04/823704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/04/823704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/1"}}},"commit":{"parents":[{"commit":"eee5156e5c9b6c18bac86fc0c5c5c68d55b619a1","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/eee5156e5c9b6c18bac86fc0c5c5c68d55b619a1"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2022-01-06 15:08:05.000000000","tz":-210},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nConflicts:\n  manifests/api.pp\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n(cherry picked from commit 1288b062e8f0ac88eea9eb6e83ba6cffd119c982)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/34733390cf1d059b59dd5431030a4febfc699d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/34733390cf1d059b59dd5431030a4febfc699d0b"}]},"branch":"refs/heads/stable/wallaby"},"4b35ca1f81ab612f42bd3322ccd4a6082f90b17a":{"kind":"REWORK","_number":2,"created":"2022-01-07 06:11:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/823704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/04/823704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/04/823704/2"}}},"commit":{"parents":[{"commit":"1168a17b79f86f26d231880bb70c898c1cb1549c","subject":"Add support for oslo_middleware/enable_proxy_headers_parsing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1168a17b79f86f26d231880bb70c898c1cb1549c"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-11-01 17:30:52.000000000","tz":-150},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-07 06:10:01.000000000","tz":540},"subject":"Add Designate API WSGI support","message":"Add Designate API WSGI support\n\nAdd support for running the Designate API under WSGI with Apache.\n\nConflicts:\n  manifests/api.pp\n\nChange-Id: I2fc87d6a4a166779e34210a377c99b313a0f1b97\n(cherry picked from commit 1288b062e8f0ac88eea9eb6e83ba6cffd119c982)\n(cherry picked from commit 9a50eb57b41fd6e3b718fbc4b7cea8e7a18ddb3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/4b35ca1f81ab612f42bd3322ccd4a6082f90b17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/4b35ca1f81ab612f42bd3322ccd4a6082f90b17a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
